)]}'
{"id":"openstack%2Fneutron-lib~873112","triplet_id":"openstack%2Fneutron-lib~master~I150e5b5660164a8f6fcc9c9161cf69e721a435ed","project":"openstack/neutron-lib","branch":"master","topic":"port-hints","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-02-28 14:21:41.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-02-23 10:18:17.000000000","reason":"\u003cGERRIT_ACCOUNT_5948\u003e replied on the change","reason_account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}}},"hashtags":[],"change_id":"I150e5b5660164a8f6fcc9c9161cf69e721a435ed","subject":"port-hint-ovs-tx-steering: api definition","status":"MERGED","created":"2023-02-08 13:46:09.000000000","updated":"2023-02-28 14:22:58.000000000","submitted":"2023-02-28 14:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"873112-port-hints","meta_rev_id":"623be4a66460f833e1815d05bf9119d606600714","_number":873112,"virtual_id_number":873112,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 14:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-02-28 12:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-02-27 02:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-02-23 10:18:17.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-02-28 12:09:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-08 15:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-23 09:58:31.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2023-02-23 10:18:17.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-02-27 02:23:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-02-28 12:09:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"161fd7449cf21b8d0f71ad4c2effe85aa0e44e61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-02-08 13:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b963b891926e431755e11b169587a0c669b2dedd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:36:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5dd23ef189644a5b40ef56964bb40d6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/217ed0598f4e47359467477b9ae14b81 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7cba64fa8245b9adfae18248cc2645 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/697b7f69ea204e9ba2e883c3f185dab4 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f3627d42e4e499caa2f24101d5aac09 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e20013213a74c7c91c0f8e6c8fca13e : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9dff151954f4255b9c6f16c3058fd5e : SUCCESS in 4m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc842c5604154abba0424ff0c48f89f1 : SUCCESS in 1h 45m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa3826e5728d4b7092416bee8865109f : SUCCESS in 3m 03s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9fc3138917ce4138ab0296486a8f4384 : SUCCESS in 1h 19m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/717704200b4a483ea610413e70b5fdde : SUCCESS in 1h 12m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5fab6798153944d9a3005c3c8cc4ecf7 : SUCCESS in 1h 02m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53a99d61c17d42cdabb1e138d22112de : SUCCESS in 1h 13m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2fbceca23f2f4be5b8794739763c1fe7 : SUCCESS in 26m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df0289ce2eb34f8fa48c6ce3800d97d5 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7b9f5b659dd57a0571e10f0671fa18ef33ee22f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-02-17 14:22:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61d8d4ff38651518cec1d91b6834f09ff52d76f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:21:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08f25e3e8fe64e83b2bf058e6a749e26\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/786e8538fe744232899ade0db5c45a78 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67cb7c58eca042a9a603fa4c4542ddff : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b797c344c39e46e988eb1145eb985669 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fed9e8fc38f467babae8c3ebb9e99ec : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac675ccd96c741e3be78bb93dda013b8 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ed64ae06c4b4ad8987ab9020219e2f9 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/def781396b0b4c90a1937884099fa4db : SUCCESS in 1h 36m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23cc4beaef8b44c2b877191e2093cb04 : SUCCESS in 4m 01s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/26519b1815694288b2a302ba997770db : SUCCESS in 1h 14m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/84d6328bd8f3424abb5db59cbd99950e : SUCCESS in 1h 32m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/913ab5c6f87141c39f0861d62ef2ed0d : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1923a8e2fd934c56b5d6fcc4bc1c04d7 : SUCCESS in 1h 22m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0db3424db20440a397a4e662ea5c8d2f : SUCCESS in 29m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/009afca762844d1e982f3017d844cc10 : FAILURE in 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"1257abb3dea298cc2e5cb7be39170cba2d5f8b53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-02-20 12:13:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8aefd80110f98b6a54e4559e7c7bbcbc07a218b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 13:46:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5928593e25d4a71b9d5a23a1a58c2d3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/348cdcaa6a2c4c0b8baa5f53ff715fd3 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23180f1611d7424fb442b9e576819547 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb51d12cc4dd4e0c9e1f3a6e1433979e : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba238c82e7e34c229f19a76701485a29 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3647f35ef3da486dbe8b296ae60ef15d : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8155a6a9a156442c8e830088cf8d55a1 : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7a0615545ba45aab6328e3c487185cf : SUCCESS in 1h 28m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6067ad4377744a6d8492eb00bdfbf804 : SUCCESS in 3m 26s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/344dddc6aaa242819d39599224778f21 : SUCCESS in 1h 22m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c1ed2a3634ee44c6b25245759fcf1bcf : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/03f3dc2232ba46d397ed6617100ce9c7 : SUCCESS in 1h 13m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a1a98eb1b614141985f28a6edefe1e4 : SUCCESS in 1h 06m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec42376360f048098d5ceb00b94398c4 : SUCCESS in 27m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d8a65d893c944499507e3dc2d68719f : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":3},{"id":"04ced58ad27320ad907448c267ff2c6e4074ca4b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-23 09:58:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d23adfbd613b9bcbdf907c7a0470cd01b9a9714","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-02-23 10:16:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eaabd8c24e592a4bfb9073a96efd250714f5df09","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-23 10:18:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ad9c88d29612e130531b37b598ed5abba4b8967","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:23:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7c87c1b99ffcc2a6889d7d99f731255291c8a8c5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-28 12:09:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"01b681ce816b90d5b5327ee66b3a1bdd754ffd10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:09:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3435f9136a202d7a421d36f284ad8ee343f4eeb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:21:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/244cbe6cfece471da49fbc3081ea84fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8dd4ad0830174c798a5c0791ac168b91 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383f1c0ed3254215921787f54fff2f66 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010859dad4ce416d92cc061cf0b783bc : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af9f40e78ec640ce9c0de7cc85bba654 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49e3b49e2a234603806891111e926e2c : SUCCESS in 4m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fba541c95a8447b78d1b0f14e773f1e4 : SUCCESS in 1h 30m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76642425ce84ef2a6c9ef05b645200c : SUCCESS in 2m 59s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/789bc48d39494a1bbd3fc048806ec1e1 : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f35a7c27bb68443fa395e224a6317912 : SUCCESS in 1h 22m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86c1a07ccfc24dde9754a1ed6d261507 : SUCCESS in 2h 10m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/317211e1f12d401ba8a616512d14e246 : SUCCESS in 1h 05m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88d336a3a4d54b02b3cf0792fcb670e3 : SUCCESS in 1h 37m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2339c0b92fe949c4b867a232239cfc21 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7b497163e129564ff4a91368ce2302877879d05f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:21:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"623be4a66460f833e1815d05bf9119d606600714","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:22:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f07740b89cc4f288f9ed774637b21e9\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6e2bba938a0f480c86390c5609391f80 : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181129dd2a9b40eea418f8e30e970495 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a955ecddf61410a8ac119b8f9501625 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ec6a39c463d0c6c6ab8cafe05c306444f8f223c5","revisions":{"a6c4331451937966731fa22102a0a71e13b01a0f":{"kind":"REWORK","_number":1,"created":"2023-02-08 13:46:09.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/12/873112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/873112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/1"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-08 13:37:23.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-08 13:44:33.000000000","tz":60},"subject":"port-hint-ovs-tx-steering: api definition","message":"port-hint-ovs-tx-steering: api definition\n\nThe ``port-hint-ovs-tx-steering`` extension allows new values (i.e. a\nhint) in the ``hints`` port attribute. It allows the control of Open\nvSwitch\u0027s Userspace Tx packet steering options. For Open vSwitch details\nplease see:\nhttps://docs.openvswitch.org/en/latest/topics/userspace-tx-steering/\n\nChange-Id: I150e5b5660164a8f6fcc9c9161cf69e721a435ed\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a6c4331451937966731fa22102a0a71e13b01a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a6c4331451937966731fa22102a0a71e13b01a0f"}]},"branch":"refs/heads/master"},"3cd9c59d4ba36aaa5914e713438ddf1e56446a27":{"kind":"REWORK","_number":2,"created":"2023-02-17 14:22:41.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/12/873112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/873112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/2"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-08 13:37:23.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-17 13:48:29.000000000","tz":60},"subject":"port-hint-ovs-tx-steering: api definition","message":"port-hint-ovs-tx-steering: api definition\n\nThe ``port-hint-ovs-tx-steering`` extension allows new values (i.e. a\nhint) in the ``hints`` port attribute. It allows the control of Open\nvSwitch\u0027s Userspace Tx packet steering options. For Open vSwitch details\nplease see:\nhttps://docs.openvswitch.org/en/latest/topics/userspace-tx-steering/\n\nChange-Id: I150e5b5660164a8f6fcc9c9161cf69e721a435ed\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3cd9c59d4ba36aaa5914e713438ddf1e56446a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3cd9c59d4ba36aaa5914e713438ddf1e56446a27"}]},"branch":"refs/heads/master"},"ec6a39c463d0c6c6ab8cafe05c306444f8f223c5":{"kind":"REWORK","_number":3,"created":"2023-02-20 12:13:14.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/12/873112/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/873112/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/873112/3"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-08 13:37:23.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-20 12:11:54.000000000","tz":60},"subject":"port-hint-ovs-tx-steering: api definition","message":"port-hint-ovs-tx-steering: api definition\n\nThe ``port-hint-ovs-tx-steering`` extension allows new values (i.e. a\nhint) in the ``hints`` port attribute. It allows the control of Open\nvSwitch\u0027s Userspace Tx packet steering options. For Open vSwitch details\nplease see:\nhttps://docs.openvswitch.org/en/latest/topics/userspace-tx-steering/\n\nChange-Id: I150e5b5660164a8f6fcc9c9161cf69e721a435ed\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ec6a39c463d0c6c6ab8cafe05c306444f8f223c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ec6a39c463d0c6c6ab8cafe05c306444f8f223c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
