)]}'
{"id":"openstack%2Fneutron-lib~874133","triplet_id":"openstack%2Fneutron-lib~stable%2Fzed~Ia0b3e39f70c30cd906ce57652f3da4c3c0adaaa8","project":"openstack/neutron-lib","branch":"stable/zed","topic":"secure-rbac-stable/zed","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-16 11:37:28.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-03-16 11:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0b3e39f70c30cd906ce57652f3da4c3c0adaaa8","subject":"Do not set system scope on context","status":"MERGED","created":"2023-02-16 21:58:23.000000000","updated":"2023-03-16 11:38:38.000000000","submitted":"2023-03-16 11:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874133-secure-rbac-stable/zed","cherry_pick_of_change":864809,"cherry_pick_of_patch_set":2,"meta_rev_id":"ccfb1ac4598ee4954d8bc61ef41c4b2e37bd6e08","_number":874133,"virtual_id_number":874133,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 11:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-16 10:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-16 10:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-16 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-03-16 10:02:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 23:41:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 09:40:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-02-20 13:39:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-20 18:05:46.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2023-03-16 10:00:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-16 10:01:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-16 10:02:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"df9b95dd099a5ce81a1a3671586310c2b53d0ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 21:58:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0655e481c66b6c64822edebab0a6a7d906e50191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 23:41:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d42eb84575640539038accb263a84a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20bf042d1bd048849cad65bdb049d074 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4839a210b7d74f40a8ac2fe6ea567131 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3121d317f0664f2ba1cae14ff9d44f71 : FAILURE in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3539f6f5b14e7fa0b75ffe0b4de2b2 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86c55b2fd8194a38bbefe8eba04e4501 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42556758c52b4b91ad9af3bb0051dbea : SUCCESS in 3m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14f9a9e5b5c945b8a73bc6a4241ae370 : SUCCESS in 1h 26m 31s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f837d33577fb49ab802cec8a36bde5d6 : SUCCESS in 1h 25m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d823816f44bd4bd3ab27af7451b93f40 : SUCCESS in 1h 40m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7418632f24d04c68a44a6a71c71efdb5 : FAILURE in 1h 32m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/708b5d04bbe24b8c92a063e9a647c885 : SUCCESS in 1h 36m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd8aded37214479b85fba2a11eb790b8 : SUCCESS in 34m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1411468697ea4741ac9717fdc44561d4 : SUCCESS in 1h 01m 34s\n\nWarning:\n  Comments left for invalid file neutron_lib/api/attributes.py\n  Comments left for invalid file neutron_lib/db/standard_attr.py","accounts_in_message":[],"_revision_number":1},{"id":"edbb889b068d5e288db72a35defe9587ea716e2a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-17 09:40:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c35f959185c4b347da3b1c55d1a0d07883504e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-20 10:52:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e08ba45eba732ff5b9a818771d8885267ca81671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:22:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba8202e4cfd044de9987dea12cf8d72e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f196972630564e7dbd8f8992702bba88 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e4021b84a604b3c866bd590ab9b08e3 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6138db34f5334b6e917975bf44bd22b8 : FAILURE in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23785e291f7f428f82f3941dc9635647 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87c3894ef8c746bdae5bc62d256842d4 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4642a341c07b46c890a833f18c0ce86b : SUCCESS in 5m 50s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3effd26d4a34a219b84b35a026743d8 : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/bed3db5e5c2642f88ea3fba33b36f682 : SUCCESS in 1h 28m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/459aa7dc8fa249dd829219bab92ef99a : SUCCESS in 1h 12m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d97539ae8dba4358aeb271c6b9d55ca3 : SUCCESS in 1h 04m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/75213a5ee6a141e691ec7da712499212 : SUCCESS in 1h 14m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ed4ac8bb0eb46e39835590f05e1f408 : SUCCESS in 41m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3257ce2144044f6abb0bf78b47e3cc6d : SUCCESS in 46m 19s\n\nWarning:\n  Comments left for invalid file neutron_lib/api/attributes.py\n  Comments left for invalid file neutron_lib/db/standard_attr.py","accounts_in_message":[],"_revision_number":2},{"id":"bd1e3191262100a0f3d6c1f860da7a270589803d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 13:39:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba6f1ef69651792294bf09cbeb13f3a4a0d62ea1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-02-20 18:05:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c270f83c547cf609a7f69a60eb6ce6b919e38149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 19:55:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5488daa24474408695d92b61d1c6a2ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c27dab12ab4744dab4b64d7b14f3c128 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e265f1c1016473bbf1661501a60572c : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/144b0a86a2a941979f101b447949f449 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ffc0087c92242d8b932a2408b0d9ed8 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16e8c0d601b441c6b117ed7d546d74d4 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc86d2d91794cf1b3f96bb3ed5d43d2 : SUCCESS in 2m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36b8497975a94aa9854e3a0571e12a46 : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/329bd5a84aad46c5aee74f46e988d314 : SUCCESS in 1h 38m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8ad9734078ee40559d87795066a9d72f : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2054618f4a743c2bb39489b80694e5a : SUCCESS in 1h 47m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4b94af95a3844616b32de1a4abf4f92b : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d8d31d7c21b43df8ff82bd3944aa3d7 : SUCCESS in 35m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8f76e9564684af48a10ffd7207cadcb : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a7af5127dc5dd3569628907f273ab6b3030b7ab5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-24 18:58:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b029a1a079c1f96cc866907cce12509f0944de81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 20:27:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4dbcaefbbe64a6aaa0c6dbe627d142b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed28617d157b4b1fa0b429aeae84cb8e : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50bb09d59f7f4489bd2d7c532157575a : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e96c2cebd824440b41bbbc30e3052d9 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08a348736c30435d917b305bf4df0d70 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bffb2e491734759b95cf795c30b6892 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ba5de3d03b144dd930959f6ca708dec : SUCCESS in 3m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e9ece914c2f447fbe88688799241b68 : FAILURE in 48m 56s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6f2af7fbe044486fb428f5704e7f1cde : SUCCESS in 1h 27m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f98c4906fc284b72b9b83ad634d5acc2 : SUCCESS in 1h 19m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b7f1b207e6b40258db1005a73196b4e : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/48f4381916564f85aa05890784893a40 : FAILURE in 17m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d8e64880ed9e430c929177df2e0b5349 : SUCCESS in 1h 13m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/33080d2dc7c143a89b3de5c7e18fabe1 : SUCCESS in 26m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/283ec346e4554b75873f11128b0b39e2 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d5d5652be9ca448863125601ff3f79bcb4319241","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-24 22:53:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff7d6ba99ee91795d7a337a31116adb333809602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 00:32:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0488c559959b4703b08f31b0950c3213\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f69a55710845faa4f66afd613ce71b : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb75aac46f86440fac7d8fdafa9896fe : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a2cf00b8264f9d9df279044b628a6e : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9288778f00648bd9507747f1cf4d995 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f00a29aa41f431a9f7d073b5be1ab0d : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5ec774c9a0f4fa19e2b7a6e47ffd210 : SUCCESS in 3m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a9ba5501c06417cbd20d96d3202d8b0 : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f43e29e1ff134643a25ec24e46fd02da : SUCCESS in 1h 35m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/64f41c534f524082842aa913ae1660be : SUCCESS in 1h 33m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7c80669820e847d9bf9699da63d990dd : SUCCESS in 1h 24m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f85e0c5eb4e4f08b56d81591c47627b : FAILURE in 15m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/300f5ea67743487a8738459b421bb407 : FAILURE in 1h 19m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77345dfe294541f19e1d3d4bb0a5a89c : SUCCESS in 30m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/053a97eb908245c6a4fafa07e46da391 : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c8f82027831c6e05562ded000728deebf1c586f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:38:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f0341954a776ca42a712b09021ca5d2ec857ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:07:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef2f6011572e48f98ba808a9464a1d8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199800d3328645709c99dec88ef75a0a : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adc36a1e400941d89b674e988e31d96a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc53f0dfa9124e9b933a7db396ab211f : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d92fd0296b4e4058969038fcd0993a75 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40c3e8c51bdf427989e994ce3f221d05 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9515abd885d54a6e9f247c51f1bcf571 : SUCCESS in 3m 18s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cb40b194ae04002af212edf1e737d04 : SUCCESS in 1h 24m 06s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b1c73951a2f343a284c416f7b7a8c9e0 : SUCCESS in 1h 20m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b9d950ce7b0f486590fc53708abbaac9 : SUCCESS in 1h 21m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7f5a349c354411090a45f6595519f59 : SUCCESS in 1h 22m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e5cbf9200cd45e9a6f75eba9bec348e : FAILURE in 14m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc402a501cb24935886c3a772773b033 : SUCCESS in 1h 28m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c09d1dfcbd5426b9824e16e3592a975 : SUCCESS in 27m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9f3eff659834bf48ef499c189415f05 : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8ec556b30fec38e110f214396a34792c9e5f6f90","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-11 01:49:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fca1f5ec338ceae266d3a25d1db2bddabd183500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 03:35:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b453a4f0c223426a86b2fe8758cb2492\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f2bbd7032542f0bc3a1dd34bb3d80c : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930a30a7e3384662b6c6911ae425e081 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d8356c287c4edca3479f38caefb32b : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aaa49c2199f401a8b611fed8f3c4506 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/806ad8e2aa494c2da8de494528deed67 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6bd2b8728764faa93d0dd04a7c786ac : SUCCESS in 3m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0525513a01994061b58a97ab893c4a3f : SUCCESS in 1h 15m 37s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/1bea06dfce284d6a846b8631de049210 : SUCCESS in 1h 40m 52s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/6fcbada1c5864fdf8057c7b6a55cfef9 : SUCCESS in 1h 31m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/afba3397b6474829aad52f3a1e6e79b0 : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c68a62a540d644ff8e79365c33472729 : SUCCESS in 1h 39m 14s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/bce0a07f6a874b79b98669c883979cfb : SUCCESS in 33m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/492ebe5847764278b0195d7c9a4abfbc : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":2},{"id":"bf1678b6777ff0bde6399f04b9e783e8088a3636","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-16 10:00:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bea8b5bcff86c8595aa1614309d24802cc37b26c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 10:01:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0d69204559229fc46d6ee0018efbd9f3a03fc74","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-16 10:02:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fcd7a767d6592be11b6056a77b3273df87d2dae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 10:03:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37e4c38a7aede3d47c108b3aefc90a7817daae61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:37:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2578cd48524d72a34dfaea2f8e6601\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d86507c98a94826a2961ed7e3814a2b : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3e2bb0bbe24e9e9ce28a5abd49a659 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342bbeeff28a424ea944250f14dfde73 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2073dd089c8d47d3b11cb04233e45f97 : SUCCESS in 3m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/088b98571d2e41bb94363abb88093fd4 : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/5ef140bffdc249918220e6b34a3826cc : SUCCESS in 1h 32m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f900ea5ea2674ead965a1bfac3bfc800 : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e8b2896f09398463810b91c859e2f39f6e10c03b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ccfb1ac4598ee4954d8bc61ef41c4b2e37bd6e08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:38:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492e485682e344c5a5ad41ed14da2e2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7491dd40dcb43209f5cb186e7543db9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bc352a4900d8b8ca42b6d4d50a2d7e104a710b7","revisions":{"f72cbfbeae04e44b58e99b6a46716488c8136c95":{"kind":"REWORK","_number":1,"created":"2023-02-16 21:58:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/33/874133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/33/874133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/1"}}},"commit":{"parents":[{"commit":"64b45c627ac164ca075542b095d34d80b12c9b95","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64b45c627ac164ca075542b095d34d80b12c9b95"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-17 02:43:48.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-02-16 21:58:23.000000000","tz":0},"subject":"Do not set system scope on context","message":"Do not set system scope on context\n\nIf enforce_scope if true then system scope (system_scope:all)\nis set on context which is not valid now because neutron does not\nsupport the ssytem scope and everything is operated on project scope\ntoken.\n\nRelated-bug: #1996836\nChange-Id: Ia0b3e39f70c30cd906ce57652f3da4c3c0adaaa8\n(cherry picked from commit cf494c8be10b36daf238fa12cf7c615656e6640d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f72cbfbeae04e44b58e99b6a46716488c8136c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f72cbfbeae04e44b58e99b6a46716488c8136c95"}]},"branch":"refs/heads/stable/zed"},"7bc352a4900d8b8ca42b6d4d50a2d7e104a710b7":{"kind":"NO_CHANGE","_number":2,"created":"2023-02-20 10:52:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/33/874133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/33/874133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/33/874133/2"}}},"commit":{"parents":[{"commit":"64b45c627ac164ca075542b095d34d80b12c9b95","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64b45c627ac164ca075542b095d34d80b12c9b95"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-17 02:43:48.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-20 10:52:46.000000000","tz":0},"subject":"Do not set system scope on context","message":"Do not set system scope on context\n\nIf enforce_scope if true then system scope (system_scope:all)\nis set on context which is not valid now because neutron does not\nsupport the ssytem scope and everything is operated on project scope\ntoken.\n\nRelated-bug: #1996836\nChange-Id: Ia0b3e39f70c30cd906ce57652f3da4c3c0adaaa8\n(cherry picked from commit cf494c8be10b36daf238fa12cf7c615656e6640d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7bc352a4900d8b8ca42b6d4d50a2d7e104a710b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7bc352a4900d8b8ca42b6d4d50a2d7e104a710b7"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
