)]}'
{"id":"openstack%2Fneutron-lib~874397","triplet_id":"openstack%2Fneutron-lib~stable%2Fwallaby~If3a97c4d3a0f4cb6b4d06434f74cbe9d933a07a4","project":"openstack/neutron-lib","branch":"stable/wallaby","topic":"bug/1987378","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-04 17:45:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3a97c4d3a0f4cb6b4d06434f74cbe9d933a07a4","subject":"Don\u0027t check context.system_scope to check project scope","status":"MERGED","created":"2023-02-20 11:01:15.000000000","updated":"2023-04-04 17:50:10.000000000","submitted":"2023-04-04 17:45:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874397-bug/1987378","cherry_pick_of_change":874396,"cherry_pick_of_patch_set":2,"meta_rev_id":"e0a7cb2fec9d0cdde09bbb00b94a116bb0f32ad9","_number":874397,"virtual_id_number":874397,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 17:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-04 12:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-04 16:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-04 16:10:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 11:19:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-20 11:50:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 11:31:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2023-04-04 16:10:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c368960cf0ec08b1e1b2e9e7bfb005a817ff51e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-20 11:01:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"59ec6a2e2cb3034981c7a0a15db4dfe31d79da47","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 11:19:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f69a3b6faa81b558fc8c994efe42f1f8878042fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 11:50:41.000000000","message":"Patch Set 1: Verified-1\n\n(61 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/707a426e7c1a49b095e122d6d4631361\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf0870e27094805923d5a25a7c0608c : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e15c1bdb25744e7db32abfdc1581d894 : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c44c7813874c06bbbd925c238d0e09 : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8734cd630ce4eb2a15ecce14aee29f3 : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b947e72f3aea4c7c940839ca44f34116 : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ae78c519fb84dbea7c184633d313d78 : FAILURE in 3m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df786e8e8f59491995b71cf408d467ac : FAILURE in 23m 09s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/62e4ec95093f443d86c7e7fa0a888750 : FAILURE in 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"05fed18d58331956e45ee99a49ebdb0010ab75ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-17 09:40:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4e03071089180b8c1ee372c44ac7427b006f75d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 11:14:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d92eee4083b4afe848d1bf4702d5482\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad3dec2b9f84193819a9fea62853e6c : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/783b29b3081944d6bc869f6035fe4aab : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb7d375d2414d108f515d332b2acb4c : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e84b2667122a4479b207bea55e03acaf : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95841d8ea5064794b833d81986b86dd8 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfda5c1a80ba49c59eb0cddb3a8712ca : SUCCESS in 3m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac0340c1181e40e78a40cffeda245f3c : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/f1b2e5a03a07469db7496030f10f067f : FAILURE in 56m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c39dbbd9cba20ec0361207d7b441bb3d4a1ee98b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-24 09:03:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58d55d2a0e95e273382f9fa1d342eb2322fbf276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 10:25:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdb3f329bd4b452cb4a1052163f1e6da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09268483cce946f68a69ffacacbf1ab4 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f85ad5dc0b94e42b03f49be1c104922 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e0b7ff3e653487297eda59d895688bd : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03c332359cff4902968aa8de2f24ccb6 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/107da730829d441fb9d2d9fbb1b1bb9a : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/691ae213c98f4cb99785ac0619d64d40 : SUCCESS in 3m 30s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3826d7faa2144989f1e39d8c0fc5d86 : SUCCESS in 1h 20m 30s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/5f74366685ba4e49912fbc622e69b6fe : FAILURE in 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"45e5ad0bd303a5cac58a174fa6cbba14013c8cdc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-30 08:02:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21f680c3ae0be6decfdf81125d1ab01f845144ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-30 08:02:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"51580b14c021a815e1037363981591cd4bfc7171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 10:15:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe5242dc51f74a9a8d5f37a4ca0c2bca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc45947c0c448b3b298a1ee50afd59a : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1d652b21494358bebfa1c115f1789b : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cf2f50c07745c8b58ffda7087017fc : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80a22c7efe024bdca40b711884f9a5dd : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c716e9e79d654dce8b73721a0b91c531 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7638edddbba24492b1ecae1bb8e64cc1 : SUCCESS in 3m 35s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1d4935ac6ed4cec8ec5719760d454d4 : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/1e660a20fb3f47ac89674b19184d8bcf : FAILURE in 59m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5e9c83467c363151986433ed54a89f7b4ef07c3f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-30 11:31:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28963ef857528c3ed964afcb5da3ccbb84550d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 10:53:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"840bdf23d8462830d53678d754809d0d9eaae3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:53:44.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"472c179165e048f8015d9cbed38c73112e35e8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 12:28:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"cf2fae0f03945cedf8ff3b187d23aed045f03de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 14:00:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/692803975b0c49dc8f6565d8d37b37fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb18f1fbc284e909788bce1b4dce642 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/804f800506034979bcf632fa4f783a0e : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c69a1438e88f434f9daea1b15af1f38c : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd7f7e21ebc24f5cb3aef96bb5c623e4 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cbf6919efe94ed3b38083baf3cd7444 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/556dc98e32074aee8634b29ae6fd57cb : SUCCESS in 4m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e208aadc07b1488ea538f65cab4c0ffb : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ad80ce788c6a4a85a230de27ff3ede13 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":4},{"id":"84abb9e4d6da960542636c2bbffe34334008dccc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-04 16:10:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"895e600614892d905721879e94722d737ae3a701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 16:11:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2ba0fbf863b966b192eff21cdf38a2d0dfc62ccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:45:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da451ddbf5914d51b597db59c37593bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678708e81b2e4261983e3dae7a6e4846 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283d39756af640888781d3784e84f9ca : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e92f8f79964d609a212cc79c17ee5e : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d2deac95dd2488aa01a1fe5f83c5da0 : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9265767472f24a9bb55619f79053afe3 : SUCCESS in 1h 33m 20s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/542266a45a304dfaa2dc1c175a6f2e39 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":4},{"id":"0e02054becf46810b302e68699c2f166a0676b95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:45:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e0a7cb2fec9d0cdde09bbb00b94a116bb0f32ad9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:50:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4436f588207f4edf8691b61b0aa3395c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980d608bf0c5434b8cbaa470220014fd : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0dadfca1ffe72f9526617c596cf76ee6f7a3fd78","revisions":{"534592fe8bb2de0f4830a5a94dd0665287598eb4":{"kind":"REWORK","_number":1,"created":"2023-02-20 11:01:15.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/874397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/874397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/1"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-11-10 15:05:11.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-20 11:01:15.000000000","tz":0},"subject":"Don\u0027t check context.system_scope to check project scope","message":"Don\u0027t check context.system_scope to check project scope\n\nDuring first attempt of the Secure RBAC implementation [1] to\nfunction model_query_scope_is_project() there was added check\nwhich was checking if context.system_scope is \"all\" in case when\nscope enforcement was enabled. It was like that because that meant\nthat it is SYSTEM_* persona context (e.g. SYSTEM_ADMIN).\nThis is not needed now as later it was agreed to have only one ADMIN\nuser which will still behave like old, \"legacy\" ADMIN user.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/781075\n\nConflicts:\n    neutron_lib/db/utils.py\n\nCloses-bug: #1996150\nChange-Id: If3a97c4d3a0f4cb6b4d06434f74cbe9d933a07a4\n(cherry picked from commit 91759b17ea5d4d6563596612208e6014dee2ed88)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/534592fe8bb2de0f4830a5a94dd0665287598eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/534592fe8bb2de0f4830a5a94dd0665287598eb4"}]},"branch":"refs/heads/stable/wallaby"},"ebaf745ac73f94f9ad04f6def33f417b40b6b5ac":{"kind":"REWORK","_number":2,"created":"2023-03-17 09:40:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/874397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/874397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/2"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-11-10 15:05:11.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-03-17 09:40:36.000000000","tz":0},"subject":"Don\u0027t check context.system_scope to check project scope","message":"Don\u0027t check context.system_scope to check project scope\n\nDuring first attempt of the Secure RBAC implementation [1] to\nfunction model_query_scope_is_project() there was added check\nwhich was checking if context.system_scope is \"all\" in case when\nscope enforcement was enabled. It was like that because that meant\nthat it is SYSTEM_* persona context (e.g. SYSTEM_ADMIN).\nThis is not needed now as later it was agreed to have only one ADMIN\nuser which will still behave like old, \"legacy\" ADMIN user.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/781075\n\nConflicts:\n    neutron_lib/db/utils.py\n    neutron_lib/tests/unit/db/test_utils.py\n\nCloses-bug: #1996150\nChange-Id: If3a97c4d3a0f4cb6b4d06434f74cbe9d933a07a4\n(cherry picked from commit 91759b17ea5d4d6563596612208e6014dee2ed88)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ebaf745ac73f94f9ad04f6def33f417b40b6b5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ebaf745ac73f94f9ad04f6def33f417b40b6b5ac"}]},"branch":"refs/heads/stable/wallaby"},"a3a957dac2dd6c63f01a1521e944a4311501bc1d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-04 10:53:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/874397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/874397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/3"}}},"commit":{"parents":[{"commit":"69faa33417c310fa78cf40bc9d1d89ab4af6b4ba","subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69faa33417c310fa78cf40bc9d1d89ab4af6b4ba"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-11-10 15:05:11.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 10:52:48.000000000","tz":120},"subject":"Don\u0027t check context.system_scope to check project scope","message":"Don\u0027t check context.system_scope to check project scope\n\nDuring first attempt of the Secure RBAC implementation [1] to\nfunction model_query_scope_is_project() there was added check\nwhich was checking if context.system_scope is \"all\" in case when\nscope enforcement was enabled. It was like that because that meant\nthat it is SYSTEM_* persona context (e.g. SYSTEM_ADMIN).\nThis is not needed now as later it was agreed to have only one ADMIN\nuser which will still behave like old, \"legacy\" ADMIN user.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/781075\n\nConflicts:\n    neutron_lib/db/utils.py\n    neutron_lib/tests/unit/db/test_utils.py\n\nCloses-bug: #1996150\nChange-Id: If3a97c4d3a0f4cb6b4d06434f74cbe9d933a07a4\n(cherry picked from commit 91759b17ea5d4d6563596612208e6014dee2ed88)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3a957dac2dd6c63f01a1521e944a4311501bc1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3a957dac2dd6c63f01a1521e944a4311501bc1d"}]},"branch":"refs/heads/stable/wallaby"},"0dadfca1ffe72f9526617c596cf76ee6f7a3fd78":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-04 12:28:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/874397/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/874397/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/874397/4"}}},"commit":{"parents":[{"commit":"0fbbc3c0abdd7910d3a02788af3df1ade48106f5","subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0fbbc3c0abdd7910d3a02788af3df1ade48106f5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-11-10 15:05:11.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 12:27:55.000000000","tz":120},"subject":"Don\u0027t check context.system_scope to check project scope","message":"Don\u0027t check context.system_scope to check project scope\n\nDuring first attempt of the Secure RBAC implementation [1] to\nfunction model_query_scope_is_project() there was added check\nwhich was checking if context.system_scope is \"all\" in case when\nscope enforcement was enabled. It was like that because that meant\nthat it is SYSTEM_* persona context (e.g. SYSTEM_ADMIN).\nThis is not needed now as later it was agreed to have only one ADMIN\nuser which will still behave like old, \"legacy\" ADMIN user.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/781075\n\nConflicts:\n    neutron_lib/db/utils.py\n    neutron_lib/tests/unit/db/test_utils.py\n\nCloses-bug: #1996150\nChange-Id: If3a97c4d3a0f4cb6b4d06434f74cbe9d933a07a4\n(cherry picked from commit 91759b17ea5d4d6563596612208e6014dee2ed88)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0dadfca1ffe72f9526617c596cf76ee6f7a3fd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0dadfca1ffe72f9526617c596cf76ee6f7a3fd78"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
