)]}'
{"id":"openstack%2Fneutron-lib~874412","triplet_id":"openstack%2Fneutron-lib~stable%2Fwallaby~I2a44e3843dd25b9db387c12f536a084d1fafbcd1","project":"openstack/neutron-lib","branch":"stable/wallaby","topic":"bug/2004538","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-18 10:28:37.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-04-18 08:54:32.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I2a44e3843dd25b9db387c12f536a084d1fafbcd1","subject":"Fix pep8 errors with pytlint v2.16.0","status":"MERGED","created":"2023-02-20 11:27:09.000000000","updated":"2023-04-18 10:29:39.000000000","submitted":"2023-04-18 10:28:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874412-bug/2004538","meta_rev_id":"a9ce6ea7314ad34c0c12e457e55eb6b383d8608f","_number":874412,"virtual_id_number":874412,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 10:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-04-18 08:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-02-20 13:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-02-20 14:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-04-18 08:54:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 12:37:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 13:24:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-02-20 14:37:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-17 15:59:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b42a8ac77a8450faaab1d26ffa2e1fd5cd46b3a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 11:27:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82ba63d14d37f379ff1ac05d933be6047e23e32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:37:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a70482649ae4706ac11362059b51f93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3a983ee0b848579dcb8b4be762b6f6 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26e6b13ccd744b43ab9b4d787856d395 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b571832ac647abb33d7c30a8c2873e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe670308b3ff4ff5ba0eb75f9059d59f : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f64470c9a740d9985d09767164ae5d : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9573adfa7e94f229adaa6f117b49ba3 : SUCCESS in 3m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2e329fdb7be4f5a815c9afc8a8668c7 : FAILURE in 38m 09s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/963a1800ac12457a80d31a6359fe86ac : FAILURE in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"efce9ec5c23bc624093919b70a2cfe2bb5140565","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-20 13:24:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a159f39970bcc16759b999feaa0e9b748e34d954","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 13:39:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea7db432fc104ced3a14d4cbcbc02b0ad45a30d1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-20 14:37:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9520dc51345bf36eee95545e46f67c133041aaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 14:50:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebcc9c38ecc645adb0a30d0a28cb41de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e7bb6890fb422281de9c623279a76c : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef9c2c56fc94770b3267c2424fab3d4 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc54b9e7766f415193cb1ee603ac5366 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f46ba6e95334a58999b1227e37a2951 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a236181b35f488b859422d18ad7bc20 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/651300732da44cafb39b8a1777e0e764 : SUCCESS in 4m 24s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48158c3e345c4c6cb70c815b04fda48c : FAILURE in 28m 43s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/f54699c993cf4ea6b7c145de9c782fef : FAILURE in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6f91044bfa67308b4d61a48e1ffe6267cbf77443","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 18:06:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcc9f501a2bcb61462edbb88f5323ad35e17e0fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 19:05:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4296f2008cc74dc98ec1d93a1157a43f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831f71640cbe45e9809f8bf6ca3ac9f1 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16fe3a45bcb24fcca3b70c3244ccb75c : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c9ab5e5f6d4a57a4f36c928ef4247e : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cb32bb5f5864fbb9c3a90a9b7e2ad8b : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41179a7df97b472c84e3f753854ea309 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8530917df180481496b1bf5ca31e31c0 : SUCCESS in 5m 55s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf87e9cd01644eb694f044f44fbe20ae : FAILURE in 40m 44s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/c318f6cc79be43c6a10938b0c5c56410 : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1b966a2b2e2be51b543aeedc28619cd353ffb2e0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-21 15:39:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"500ff0dc1306ca33e4ad32968f5b40a01f48f5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:42:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ac077af54f49d9b1bf09fa410bb11e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8147b56054040e5a4ea18d8f54ab5bc : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b668b98f9f6b4694949cb2f600b6e303 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d55c1ff7ae4eed8c04ec02db73cd24 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38295864b29541ea89a536b6427f74fc : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7026fc3893244b208a109503624456ff : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5337260c464c4395939cf2c5647c8549 : SUCCESS in 3m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d580fd139899451985c72173025ffa72 : FAILURE in 37m 41s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/55a91de983cd411a8610dcef5d00f7b8 : FAILURE in 54m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e9dcfa613bcfe683de3cd82c74576a14005d48f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 11:00:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65a6d6b974065d3115e6acf727b0e7eed5114f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 12:31:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a855f8fcc0d43c3bcbc01938625ea8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c5f5d73c7644999ee1cc30fde2c686 : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24b5e79b73554964b565de9b51af6dc0 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45218115f45a434cb338525bd3893a38 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f431d5e72e1f4a2b88368581cb925609 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0feb7d62f20e48719caa972ad50b42c4 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3ac31126ba34ca494fe849e1ccd76bf : SUCCESS in 3m 25s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b258a257ca14b40a29b6ab229366bac : SUCCESS in 1h 26m 02s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/2cbeb71e1c8d40c698b5b7792f168d22 : FAILURE in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5c393eff2817ba533497c5de9db65aa15b75a8df","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 14:57:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d35a41dc319a1a25e69d6a57e834f01afeab5b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 16:15:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15bf81d6ca7e4942a99987dd1b1b4e79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb048e75a7d040f58e2fc6d1340ba53b : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35f5322b48ea46b288056b56a8c3676c : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623394c5565645868d0e580f37e6cf1b : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33f773edee4b44059ac862cad6396a7a : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55dd2f3a53a6481b8bd2e93a22733455 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad938105cfcf45b9bec079b6ece9c0cb : SUCCESS in 3m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c66e3dda1204a0492d105266bf41685 : SUCCESS in 1h 15m 58s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/cb94a196c789451db7175e8a39730aa3 : FAILURE in 56m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d9141c3ea8defc97530012a02f54a5da05bf3532","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 10:47:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eac62ddb386dc0464074667d8ca16ed28645d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:33:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/646fcabd295340c396549b499e8111a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1b4d5ad0b14a35a609cad89f6b29fc : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eacbebbb621c4e909185d83de920b381 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d01df20ee664c009cf65a6629c45dab : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f4cecf3ec9c452f99f3c7547ea84879 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29f2cb48693e4be39150651d118ea3c8 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9db602016bc6499695de4a94a49c1278 : SUCCESS in 3m 20s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f30585a051c4dc0af3c5fd8eec0703b : SUCCESS in 1h 35m 55s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/eea0cb55720244108db9fb1ffd10fcbb : FAILURE in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"883bca026cc5a52dca7e82d7a92374ff04545889","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-24 14:16:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee35e70344da38629f2ed7d2d043fc3d9b97e5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 15:59:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/068fbf6acf56467cb802c79b89723fb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e19615e79a6484a8f230a808462e27d : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1d4cff67c294dfd8562a916a0dae771 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5832422c0034473d9db548be8bf66d8e : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4554214382434b74a77969dc0a7883ff : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f4647f17f02497cb7d892c7afb6abe8 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf0d9ae00ce49bb97b1f913da0a21f8 : SUCCESS in 4m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/746a9ec749c6408abf8bb5e4492f08cd : SUCCESS in 1h 35m 19s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/16f3da19ee144083a0a563620f79986c : FAILURE in 1h 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a5df264c476aadbe0a96fd86a18491bdec6cc5e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-27 09:28:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4610ab8cabfc9a011728cfa04ff37fbc3c2d777","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 12:58:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0718375e8646b1e7f609729073fa53791431d76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:12:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/575aab354035468a995e8d3c65d437ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c44f8962d8841e081183ba71302df62 : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6387eb75bb8c487cadcb7d77b19500be : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b71c19e8174fadaa5157af3f30dd87 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2280d48c1b34fbf8acc00171eaf9025 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bda4cc52f654a2bb5dd3bbf7431dc25 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab05339b254c43699b928add6b24fcdb : SUCCESS in 3m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64de9d3f1ae74d5888350590413a148e : SUCCESS in 2h 09m 32s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ecca49cff3904e7f9b1a67b69b109a5e : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":1},{"id":"08a62502e6a6173fa2fceb03883798c04d02b8eb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-18 08:54:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84e41cabe7e7cbad7db6e039e1983c4eeaf3a59f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 08:55:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42ab60ca290c03f2e82e32307ee51213c0f17ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 10:28:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5452df8afa634960a6b36c9872129511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f459aeba64294d69a542b770b63ea7b7 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777cc680bd79451683fcae35b0244ab0 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80f1a23e745d4d9b9ff8d2071ebb3e04 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d4c671df8d748dc9b2e3a3e32e4dc09 : SUCCESS in 4m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b25f6d4c2cde4c6aaa91fd9856029f00 : SUCCESS in 1h 32m 25s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/dd75311cd47e4774b7dbea64d46d7f94 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3a27ea5dbc865fd3d8185d5ff0128da33700960e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 10:28:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a9ce6ea7314ad34c0c12e457e55eb6b383d8608f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 10:29:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16b14f3123942808a40fc849782007b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e9a5d5138b4b58b081260c877bba62 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"025e0d72eebe81dcf8ba931e78bfdbe72ecd122b","revisions":{"025e0d72eebe81dcf8ba931e78bfdbe72ecd122b":{"kind":"REWORK","_number":1,"created":"2023-02-20 11:27:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/874412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/874412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/874412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/874412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/874412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/874412/1"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-02 09:34:44.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-16 04:13:54.000000000","tz":60},"subject":"Fix pep8 errors with pytlint v2.16.0","message":"Fix pep8 errors with pytlint v2.16.0\n\nChange-Id: I2a44e3843dd25b9db387c12f536a084d1fafbcd1\nCloses-Bug: #2004538\n(cherry picked from commit b1a4f783daa1c48057cb450e58198f1bf93931c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/025e0d72eebe81dcf8ba931e78bfdbe72ecd122b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/025e0d72eebe81dcf8ba931e78bfdbe72ecd122b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
