)]}'
{"id":"openstack%2Fneutron-lib~874866","triplet_id":"openstack%2Fneutron-lib~master~Ic4d7a0740f7136cac2055e57e57f66e76cb150f3","project":"openstack/neutron-lib","branch":"master","topic":"bug/2008223","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-28 11:30:29.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-02-28 10:04:04.000000000","reason":"\u003cGERRIT_ACCOUNT_5948\u003e replied on the change","reason_account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}}},"hashtags":[],"change_id":"Ic4d7a0740f7136cac2055e57e57f66e76cb150f3","subject":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()","status":"MERGED","created":"2023-02-23 10:24:55.000000000","updated":"2023-02-28 11:31:35.000000000","submitted":"2023-02-28 11:30:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874866-bug/2008223","meta_rev_id":"bfe8db2499b88fb210518011811200ac9d0d8fb0","_number":874866,"virtual_id_number":874866,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 11:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-02-28 09:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-02-28 10:04:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-02-28 10:04:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 12:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 08:12:16.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2023-02-28 09:41:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-28 10:04:04.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"cba5a25fe459484910bbd1ef5b1d1b83670aa75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 10:24:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57ebd275a1e8bc43a47b73741c789cdf882f0633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 10:28:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbefaf34129b70ed83dc75ccee3ebed47883f245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 10:35:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"240bf249a95d05dd6d92dbf678a5a843bf176539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 12:28:08.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a71146993c64fdd959e7ba7fc09dc81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7051a43cf55f49a3a6b5d6de96348ab3 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d949ba9dd94c9fb4892dc50da0a14c : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548493395c054b2b9589bb5e3906c3c7 : FAILURE in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8582b4e3b4b64848b40def2a3b1c6aaf : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80debc199e7b4e729b2de324a7dba240 : SUCCESS in 4m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/286f9ac75d924b65b123ab1ad6adc7cd : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d047ad598a924e238e8f3619c3b93a18 : FAILURE in 1h 50m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9afa6c4b3bab40a1b91169b4fc6ee819 : SUCCESS in 1h 43m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28a899ad44a0412f8ab5df96cb3aea3a : SUCCESS in 1h 07m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/320516098ea64cd5ad8e3d60a93645b2 : SUCCESS in 1h 10m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e7e3ce8d20974580aa83b31ff6cb6b6d : SUCCESS in 1h 10m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/352097f4b26c435db6a8c68de027f628 : SUCCESS in 26m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f612038f4e2433aa3808a6d8d1899a3 : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":3},{"id":"51298ea90caec2e5bc083849e0201e4a8238288e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 14:57:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f6d0d188cc71b7dd27bbf14b20076456ccf42085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 16:55:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1692551244314b02b1ee891c9ee179df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100ca65ea7994661840b05a403a7f3af : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aa271e63cdf4c009d666f2f3fe1f0bf : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c36080dd7714d76971a6544d81a073e : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b0e6721f354beb91c873775414f920 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a636781445cc45bda078a7b1fd8c37fa : SUCCESS in 4m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32e10f4616b24dd28b7f8730c77cdd33 : SUCCESS in 1h 20m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/472cabf3c9e2424eab6ef20acb961912 : SUCCESS in 1h 31m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/145bd3ede894423da6a5444a209cb9dd : SUCCESS in 1h 46m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f67b36237a54006a5102d8dec7df55c : SUCCESS in 1h 50m 28s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1543c573d8b14866af418a5e9b1828a0 : SUCCESS in 1h 55m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a8e6c72120714125a24adb9d3e944a96 : SUCCESS in 1h 30m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24bacb657956412c888c203d102791fe : SUCCESS in 39m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/850e689ada444302bd06e178033f7dec : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":4},{"id":"ea5b2a999e6e99f398d0e66caf5d503802a8e4a4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-27 08:12:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c31b5c5626f25942ba31269f6fbd8ae57734500","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 09:05:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d64bb8700444cab1d8ece330e2914c356918f79","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-28 09:41:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7ee0dd954bfbb928b58aa226ceb630ab38fc5217","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-28 09:42:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd9a49b3c04a9de15c1cfc68bf7d791440e68832","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-28 10:04:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9e43be20bb069acbeea2d98fd47e803bf00a11c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:04:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1710b6273aa58f79bcf4759eec0ee3e73b06a268","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:30:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c174925cf8134fa78503abd220b526fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb67b4718bc442e946c85ec456d57d6 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ca3e32ced14a03b7777fe066459789 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01fa5aa5a73447069fcad8a791b30e07 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b1ecdb6cdda46cb83d3c75cd8d494d6 : SUCCESS in 3m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a6948b4a5004913aaf62c92d3c0bb3f : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/154a54d8ba60448a9d67309bf53b5a8b : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2a9a17714054443c8369aad0289c0700 : SUCCESS in 1h 14m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4330c5bf4b064d2380ce501f08ff0ab5 : SUCCESS in 1h 13m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bbda1cb7d30d43e287bb41c88cbb3207 : SUCCESS in 1h 11m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b59841eae4c74e7ebdde59ce6eae4d8d : SUCCESS in 1h 12m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d4e8c7c612b43e7bca32c64ce610229 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":4},{"id":"be12250b17381b58dd70bcd6cc2df8ef02a67d93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:30:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bfe8db2499b88fb210518011811200ac9d0d8fb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:31:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96c8e52d5a644024ab745e6d2c5a5754\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a397116ead674ac4a6ada42a213b3de9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"444cec2ba04d12980b116086b727c7ff40db63f4","revisions":{"bb2cfb4eaec6c90b86eba0d76ee447c1e9402ed2":{"kind":"REWORK","_number":1,"created":"2023-02-23 10:24:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/874866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/874866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/1"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 13:50:58.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 13:50:58.000000000","tz":60},"subject":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()","message":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()\n\nAccording to the SLQAlchemy documentation, \"the Engine.execute() method\nis considered legacy as of the 1.x series of SQLAlchemy and will be\nremoved in 2.0. All statement execution in SQLAlchemy 2.0 is performed\nby the Connection.execute() method of Connection, or in the ORM by the\nSession.execute() method of Session.\"\n\nChange-Id: Ic4d7a0740f7136cac2055e57e57f66e76cb150f3\nCloses-Bug: #2008223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bb2cfb4eaec6c90b86eba0d76ee447c1e9402ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bb2cfb4eaec6c90b86eba0d76ee447c1e9402ed2"}]},"branch":"refs/heads/master"},"a966b22140080a007e9685676a0918faaed9ba6d":{"kind":"REWORK","_number":2,"created":"2023-02-23 10:28:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/874866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/874866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/2"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 13:50:58.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 14:01:15.000000000","tz":60},"subject":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()","message":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()\n\nAccording to the SLQAlchemy documentation, \"the Engine.execute() method\nis considered legacy as of the 1.x series of SQLAlchemy and will be\nremoved in 2.0. All statement execution in SQLAlchemy 2.0 is performed\nby the Connection.execute() method of Connection, or in the ORM by the\nSession.execute() method of Session.\"\n\nChange-Id: Ic4d7a0740f7136cac2055e57e57f66e76cb150f3\nCloses-Bug: #2008223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a966b22140080a007e9685676a0918faaed9ba6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a966b22140080a007e9685676a0918faaed9ba6d"}]},"branch":"refs/heads/master"},"dba5fcc972fe523ae4036a705cdacb0c3a20cd03":{"kind":"REWORK","_number":3,"created":"2023-02-23 10:35:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/874866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/874866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/3"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 13:50:58.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 14:07:42.000000000","tz":60},"subject":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()","message":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()\n\nAccording to the SLQAlchemy documentation, \"the Engine.execute() method\nis considered legacy as of the 1.x series of SQLAlchemy and will be\nremoved in 2.0. All statement execution in SQLAlchemy 2.0 is performed\nby the Connection.execute() method of Connection, or in the ORM by the\nSession.execute() method of Session.\"\n\nChange-Id: Ic4d7a0740f7136cac2055e57e57f66e76cb150f3\nCloses-Bug: #2008223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dba5fcc972fe523ae4036a705cdacb0c3a20cd03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dba5fcc972fe523ae4036a705cdacb0c3a20cd03"}]},"branch":"refs/heads/master"},"444cec2ba04d12980b116086b727c7ff40db63f4":{"kind":"REWORK","_number":4,"created":"2023-02-23 14:57:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/874866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/874866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/874866/4"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 13:50:58.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-23 14:57:37.000000000","tz":0},"subject":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()","message":"[sqlalchemy-20] Use Connection.execute() instead of Engine.execute()\n\nAccording to the SLQAlchemy documentation, \"the Engine.execute() method\nis considered legacy as of the 1.x series of SQLAlchemy and will be\nremoved in 2.0. All statement execution in SQLAlchemy 2.0 is performed\nby the Connection.execute() method of Connection, or in the ORM by the\nSession.execute() method of Session.\"\n\nChange-Id: Ic4d7a0740f7136cac2055e57e57f66e76cb150f3\nCloses-Bug: #2008223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/444cec2ba04d12980b116086b727c7ff40db63f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/444cec2ba04d12980b116086b727c7ff40db63f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
