)]}'
{"id":"openstack%2Fneutron-lib~875305","triplet_id":"openstack%2Fneutron-lib~master~I32830f4dddbb5a1a5bce43bbe47d10e4e83ff1ca","project":"openstack/neutron-lib","branch":"master","topic":"bug/2007414","attention_set":{},"removed_from_attention_set":{"32667":{"account":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"last_update":"2023-02-27 16:11:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I32830f4dddbb5a1a5bce43bbe47d10e4e83ff1ca","subject":"Remove unused rbac rule","status":"MERGED","created":"2023-02-26 03:42:14.000000000","updated":"2023-02-27 16:13:11.000000000","submitted":"2023-02-27 16:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875305-bug/2007414","meta_rev_id":"acdc6fd73beec03953f16870ca9f464ed7bf326e","_number":875305,"virtual_id_number":875305,"owner":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-27 16:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-02-27 14:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-02-27 09:33:27.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-02-27 14:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-26 05:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 09:33:27.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-02-27 14:17:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c313a9d953c409e044215becb05e60a516f82839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2023-02-26 03:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de23328c902566a042c38fb9502cacf2306d3c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 05:03:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/116457d6041944138c9749f5fee45690\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334abba601474ce9ba04f475835aec7c : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cc0d203a5784dd3aee94f475180d6e5 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03363fb2d604b53a83c29e0a373b063 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd03b1a773b0455b89e8961841acbac5 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04ceb42d23e640aeaeb9dd1c34490e4b : SUCCESS in 3m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc647a13c4e64817b7c3fbc0942ba73f : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e96cb16496174479b92b28ea927d011d : SUCCESS in 31m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b6313db4b054421998a099d2ad822816 : SUCCESS in 33m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47118f7743e2443d95c55471736a0900 : SUCCESS in 33m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/34d58447178341f1b2a75f50167e8ef3 : SUCCESS in 33m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9fc1a079c1f9405cb2a122429a7069b5 : SUCCESS in 30m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e56db016938c4d8a80ad93b12991e660 : FAILURE in 33m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/484a99e68ff841598066ea5b383b95cd : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e8a0166fe51cc74c1b6bd2913875916cccda210b","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2023-02-27 01:48:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08b4257fad9c30af3e872eafdf65280e586a7a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 03:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d625703b4764461cb6b40ef81c3f1864\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d0f09103e24498a3c7e65400e48cde : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042255a7a5a94723a55567643e316866 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fded7de21a44656939ad6e45b6b0a79 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f36698972b3a48ebba1069c205feb0a6 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5f38863fa714b11953b933545f44843 : SUCCESS in 3m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72079c9156bc4ce38523f8f908a5eb5d : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/72efaee541104da4ae1b47b9196b376e : SUCCESS in 1h 42m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f42026737bef4ac6ba7c9f4e32c0848a : SUCCESS in 1h 32m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/769ed61d0482432089733b0532865a67 : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f6a99dbf440641a497d5228279d3184e : SUCCESS in 1h 27m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f34596f2b90e4c1685e0026b53a402ac : SUCCESS in 1h 31m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/125f66e4305447518b154d873f58151b : SUCCESS in 34m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1cbe7b006e7c4113b415732832423488 : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c171d943d0a9e9f32987b918dd7a8161c589761f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-27 09:33:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c0da9ab5f6dd7765788f568cf5283da80c69900a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-27 14:17:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9d61eba166142421da6fa0858d4164ad63e533e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:18:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d26e97c0176787d816635a62b1e28414d4e25c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:11:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52c3f5ebc50741b39c8805b77d720521\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/211147b3dd5f453da2cb9f18e4b81ab7 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a2c71d1bb740c28e75ae4dd1de28ce : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d25a4ae835e4d79826ba10cf5104b35 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73729e58ba224f06bc13d8f4424ccc06 : SUCCESS in 4m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e82c74d80eb4bb69baeb824af8aba58 : SUCCESS in 1h 09m 09s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3f4df15577e1449498d4a089cd0491d1 : SUCCESS in 1h 51m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4a3e96b46954102875a2123bd1fee46 : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fb8cbd814c83400893ce597b9b85d146 : SUCCESS in 1h 48m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67795ac8d81a4b7f88a6dea908421a68 : SUCCESS in 1h 39m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb82b2caf0294dac9032138fda50bf34 : SUCCESS in 1h 37m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5510c2dcf2ca41aa88a18cff98550dc6 : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c41b8ce9f1c8feac7b2fe68578092e4b7d347b74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:11:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"acdc6fd73beec03953f16870ca9f464ed7bf326e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:13:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa94214567454b979e273b62afd6928e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e51b4f69ce43cda08980a96f2b4fda : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf5b293a552c139e37acafd73b0e856ddf01dd79","revisions":{"cf5b293a552c139e37acafd73b0e856ddf01dd79":{"kind":"REWORK","_number":1,"created":"2023-02-26 03:42:14.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/05/875305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/05/875305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/875305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/875305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/875305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/05/875305/1"}}},"commit":{"parents":[{"commit":"23fa6b3c2f46be6de8523355c786a682fd4419ef","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23fa6b3c2f46be6de8523355c786a682fd4419ef"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2023-02-26 03:40:27.000000000","tz":480},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2023-02-26 03:40:27.000000000","tz":480},"subject":"Remove unused rbac rule","message":"Remove unused rbac rule\n\nCloses-Bug: #2007414\nChange-Id: I32830f4dddbb5a1a5bce43bbe47d10e4e83ff1ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cf5b293a552c139e37acafd73b0e856ddf01dd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cf5b293a552c139e37acafd73b0e856ddf01dd79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
