)]}'
{"id":"openstack%2Fneutron-lib~879042","triplet_id":"openstack%2Fneutron-lib~master~I91fdcb93b4633d59c4c2424567fb01eee3c2571e","project":"openstack/neutron-lib","branch":"master","topic":"port-hints","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-04-05 15:10:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15554\u003e replied on the change","reason_account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"}}},"hashtags":[],"change_id":"I91fdcb93b4633d59c4c2424567fb01eee3c2571e","subject":"port-hints: api-ref: Add field length limitation","status":"MERGED","created":"2023-03-30 12:20:59.000000000","updated":"2023-04-05 18:54:39.000000000","submitted":"2023-04-05 18:53:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879042-port-hints","meta_rev_id":"2c5f7ad80abcad08a973c1b1fc2701f3e7b5329a","_number":879042,"virtual_id_number":879042,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-05 18:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-04-03 07:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-04-04 07:53:11.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-04-04 07:53:11.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 13:59:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 07:36:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-04-03 07:36:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-04 07:53:11.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"2973b898af0da83f8b3f1411b090ff24996e52cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-03-30 12:20:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1554c53c2c8eaa3dafc56dbfa2ab615cad48cbe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:59:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0460378a98d74a7b87d9902cb0f14c96\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84a1f95076fe42359d1256752c91e263 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24456cf0b4924281b83f618f68a6054c : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfad81be7aa64aff9ba7ff070bff4ef0 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/26a08e1242244206a4fbd2f20ca3a430 : SUCCESS in 1h 37m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9a408898e1344df89acbdf9d602a66fa : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a175bcd77894f1393f337b8cf1220d1 : SUCCESS in 1h 22m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/78e1768a7ee4477dae7284903e53f5cf : SUCCESS in 1h 26m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6991d65bbb3e40649a56f8b58b33e597 : SUCCESS in 1h 22m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac316872db46415a984b37ceff664bc7 : SUCCESS in 28m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f630d3834ab3433f98056468b2d78861 : FAILURE in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e0addb1ae08b9e2f10d38b0738ff462b5a4833ab","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 07:36:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8484d5f6c0efda8f314c7dbbff224af8765be901","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 07:36:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5053fcd829c23f1d9eb70d6fb6a68025aaad971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 09:13:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0698acf192bc43e8bba80f34600a9f50\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9f78b91260d441288aabb34708a0306 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fab5b83667a43fe928c9786d5b27940 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9b7b18d139461eb021e9f47c9b36b9 : SUCCESS in 4m 10s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4b89f8b2ce52457dbb3d272c3a6d893e : SUCCESS in 1h 22m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/61f0a75072df4de88b34330b5b269c65 : SUCCESS in 1h 20m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/862fb33ddfec4cfe8895ad74f1052b3d : SUCCESS in 1h 35m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/325a1546c7f34f1b921b7d4818d4ff75 : SUCCESS in 1h 23m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f7e90f9c8f80449c920377c6f195b185 : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f605f3a50fd445bdab8b3cdd93a445bb : SUCCESS in 27m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6b663a590514de081e2b21a20c556e2 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":1},{"id":"123040342afc82e22073ffe06c4989d278229c91","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-04-04 07:53:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a17be5bfa626aa875fd2f46633e4ece73708810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 07:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3a7187c7bb697ce648811857b3f7a762abbe355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:17:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ba5e09c906e44c1a38e1654ee2f4102\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c69f6cb01d84df1842b88288d25dabf : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3be2fe70d14ffcba14713f0fe22538 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2d92b7a06346dd811440b9e4e60d21 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6c16c3e4a1bb44b397a1ffbf2039585a : SUCCESS in 2h 15m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c33ee7733483444c9bcdfbba58ca49b5 : FAILURE in 6m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7312833d29664b9984cb1d1ca2323188 : FAILURE in 5m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/051b5a79f6e64fac97c0fb694994a576 : SUCCESS in 1h 59m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a83e42dabe174d91bdb477b8f6604178 : SUCCESS in 1h 51m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38290221654b4776ba1fa128e301686c : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3750585872e36e685ba76fba361920d7ade6362c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-04-05 15:10:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0df6ac3ec80239b6e802d31838f06983b56ba27c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:16:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/016a5626c5c64df5ac77d39463bd3c9c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/abbec653f91840c8824c6062fe682123 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f286f610ae64ebda752c38e98cab94f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2955ab1399d841daaf785ba41ba238e2 : SUCCESS in 5m 19s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5b16f2738ebd41d880d74e12c4bd08c1 : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/856f0982574640338bc0a465f36fc1a6 : SUCCESS in 2h 01m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f65a73564398454b81d876fcc893054f : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0203d32e11b94ee6acf3a50b85d36a3d : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce1a8f5f47174aaa9e01196831e857a8 : SUCCESS in 1h 35m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e60745c3134147169a1ecf0790cc3480 : SUCCESS in 38m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9987bde7a06540a38031547b5b052352 : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fc3ee3358e87b9d92abb2b314ea46142d7bda208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:17:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd25856eb700d3951e18c67594860b01ae9e6bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 18:53:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b83f640b64464c1caef98d005608124c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e476c757ab94fd78265efe736d153d9 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb2e179eb104d9982d07c86a2ee92e6 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9f017a24584d21b7466b581d2c85e6 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/625dc7f33f5e41f688867281d02e0a59 : SUCCESS in 1h 35m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c0d530715ef14362b05f57b3b101bc7c : SUCCESS in 1h 12m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/96465c2b78544f1d9bb4b120accac3e9 : SUCCESS in 1h 20m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0617f2a1ffa64d18a516f5cb7a3b816b : SUCCESS in 1h 24m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4ecbf8187e8a4268a3ef7acba2a84e70 : SUCCESS in 1h 27m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2223f75eb97d4db6ac33547edadbd86b : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"df43b22caf5480d5015fb362b68934bac7eb694d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 18:53:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c5f7ad80abcad08a973c1b1fc2701f3e7b5329a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 18:54:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee9dc8e6121f4ccfa20e103d2d0a0453\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/30dbb643894d406f82c401ef8fe5fa76 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a3d907a2c045688be5c2c31e21fbdb : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08cdc726b916a0a71418d45ce96364d2994d730a","revisions":{"08cdc726b916a0a71418d45ce96364d2994d730a":{"kind":"REWORK","_number":1,"created":"2023-03-30 12:20:59.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/42/879042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/42/879042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/879042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/879042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/879042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/42/879042/1"}}},"commit":{"parents":[{"commit":"ad8bcd2b3ae568aa7b8bca5b6c2048b5ba1e7d2e","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ad8bcd2b3ae568aa7b8bca5b6c2048b5ba1e7d2e"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-03-30 12:18:51.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-03-30 12:18:51.000000000","tz":120},"subject":"port-hints: api-ref: Add field length limitation","message":"port-hints: api-ref: Add field length limitation\n\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n\nChange-Id: I91fdcb93b4633d59c4c2424567fb01eee3c2571e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/08cdc726b916a0a71418d45ce96364d2994d730a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/08cdc726b916a0a71418d45ce96364d2994d730a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
