)]}'
{"id":"openstack%2Fneutron-lib~879424","triplet_id":"openstack%2Fneutron-lib~stable%2Fwallaby~I2d54ca18196765b118eb34f9721fa696980b83b8","project":"openstack/neutron-lib","branch":"stable/wallaby","topic":"bug/1987378","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-04 17:45:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d54ca18196765b118eb34f9721fa696980b83b8","subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","status":"MERGED","created":"2023-04-04 07:02:50.000000000","updated":"2023-04-04 17:45:28.000000000","submitted":"2023-04-04 17:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879424-bug/1987378","meta_rev_id":"c74fe6f4bea12e95fb0c4d4f9f402c6655d06af3","_number":879424,"virtual_id_number":879424,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 17:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-04 13:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-04 15:14:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-04 15:14:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-04 08:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 13:05:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-04 15:14:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"60a4a5fdc5e4b7b09d407b929f4851c4231ffec2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 07:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f80912735d4d6a7d5a3cc12c6d83b4696d23e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 08:12:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5efa801f5aa746f5b71fbf46f2a86103\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e494268e3e5844c6840878aed9de7874 : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fca8a819c0014c65af5b872941f4a46e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84ee16b19bf40d19a7e24e920172e02 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dad5ea9fb3e43fa96ef80cb3a383831 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67f243f39dd44d35bba6d2dcfd226f4d : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3074d2a04ac346c5b6fe947389789042 : SUCCESS in 3m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ee7b283b4b742d8bfa8ed732beccd40 : SUCCESS in 1h 08m 20s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/d9dd7cc861dd4b468b4eab0d19dd29b6 : FAILURE in 53m 18s","accounts_in_message":[],"_revision_number":1},{"id":"02df8d5cc79d2901e984791046a58fae697c63c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 09:07:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"98553e3dc2ce99a5326e8af52445840a8b1a1be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:35:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c0123da4a5b4713a6447dbfdb53a316\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffba067f54b4427698280744fc9c7be1 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff5c65f054344bf4be242b7a0a15769c : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79774bd1b04e4c59ac88eabdbd10a578 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e2432df8baf4578bc06afe6528ed942 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48a359486e9a483599fbd2326fbb0a3b : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4089890320a042c4ae3a95ffc978b8db : SUCCESS in 4m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba31e3731c27487e83a612e9216d05af : SUCCESS in 1h 21m 53s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/88debe5c1592443f97badf13d52e544a : FAILURE in 59m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c75722191f5642b443a9206b0d02712ecad2c867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 10:53:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e6053c7f6ab63d0df9f6bd881d2936b49bc9bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:53:38.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"499884542c9c1d314ede38dff4a470f2ed4ae8cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 12:28:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"75c406bf2ed1a4738879e2cf5d0874a960fa54ec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-04 13:05:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"199ea0cd2153c281a890917ec9b9ae4a126ee134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 13:47:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aeb7af095e74dc0afacf7d83ef8f53e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df970364da464f29b20f7e04455bd37d : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f541c1d09bc4e349ddbfecbdab0d3b2 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18463b7f0a2840c59e27570558924b79 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/253fdd79b8764e03be454e9a5b1eaf54 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7696fe5b4d474d0ab6fa580029776e3c : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef528b1f90f740878eb61488594a6d96 : SUCCESS in 3m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c44bc3f163e94385b0db3a40eb3d10d6 : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/4ca99a8b8c854ac38aa079c673c2f609 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":4},{"id":"54ff08ad2d353da48cb52b5ebe6a2f04378881af","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-04 15:14:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5877a8a1ce56b44b805e790848497dc03a894ee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 16:10:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e311c1e9110d14a7ed3a2cd1aabb849c5b4535a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:45:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be422c25a1b4068ab36aebe13345739\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4951a7c43dd7495ba641b82f9744bc45 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd9df53bfd04585b9564023305986fa : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b9329eb8a154202b1afb3d13491ae21 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1914f44621546e685934c2f3425d463 : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8de762b28b4c4755bf94041162ea5e93 : SUCCESS in 1h 27m 26s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/b46947ec7de74d8f925aef0dda4f58d9 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c74fe6f4bea12e95fb0c4d4f9f402c6655d06af3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:45:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0fbbc3c0abdd7910d3a02788af3df1ade48106f5","revisions":{"9997126540407fce0e51f5c505af1578e9d58644":{"kind":"REWORK","_number":1,"created":"2023-04-04 07:02:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/24/879424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/879424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/1"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 07:02:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 07:02:21.000000000","tz":120},"subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","message":"Add neutron-lib to required projects in neutron-tempest-plugin-api job\n\nChange-Id: I2d54ca18196765b118eb34f9721fa696980b83b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9997126540407fce0e51f5c505af1578e9d58644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9997126540407fce0e51f5c505af1578e9d58644"}]},"branch":"refs/heads/stable/wallaby"},"7a515a54504f49c737e55b16ceb26f5ad38c4eb6":{"kind":"REWORK","_number":2,"created":"2023-04-04 09:07:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/24/879424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/879424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/2"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 07:02:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 09:07:01.000000000","tz":0},"subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","message":"Add neutron-lib to required projects in neutron-tempest-plugin-api job\n\nChange-Id: I2d54ca18196765b118eb34f9721fa696980b83b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a515a54504f49c737e55b16ceb26f5ad38c4eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a515a54504f49c737e55b16ceb26f5ad38c4eb6"}]},"branch":"refs/heads/stable/wallaby"},"69faa33417c310fa78cf40bc9d1d89ab4af6b4ba":{"kind":"REWORK","_number":3,"created":"2023-04-04 10:53:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/24/879424/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/879424/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/3"}}},"commit":{"parents":[{"commit":"21ef9eccb93a99292941cd4667103ec572839771","subject":"Skip test_qos_dscp_create_and_update API test in stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/21ef9eccb93a99292941cd4667103ec572839771"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 07:02:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 10:52:22.000000000","tz":120},"subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","message":"Add neutron-lib to required projects in neutron-tempest-plugin-api job\n\nChange-Id: I2d54ca18196765b118eb34f9721fa696980b83b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69faa33417c310fa78cf40bc9d1d89ab4af6b4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69faa33417c310fa78cf40bc9d1d89ab4af6b4ba"}]},"branch":"refs/heads/stable/wallaby"},"0fbbc3c0abdd7910d3a02788af3df1ade48106f5":{"kind":"REWORK","_number":4,"created":"2023-04-04 12:28:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/24/879424/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/879424/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/879424/4"}}},"commit":{"parents":[{"commit":"7ed946347c471e11bdc43ee117782e892de169f6","subject":"Skip test_qos_dscp_create_and_update API test in stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7ed946347c471e11bdc43ee117782e892de169f6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 07:02:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 12:27:42.000000000","tz":120},"subject":"Add neutron-lib to required projects in neutron-tempest-plugin-api job","message":"Add neutron-lib to required projects in neutron-tempest-plugin-api job\n\nChange-Id: I2d54ca18196765b118eb34f9721fa696980b83b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0fbbc3c0abdd7910d3a02788af3df1ade48106f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0fbbc3c0abdd7910d3a02788af3df1ade48106f5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
