)]}'
{"id":"openstack%2Fneutron-lib~879441","triplet_id":"openstack%2Fneutron-lib~stable%2Fwallaby~I16ec53356ec3c5daa1dfa192a1cfa01816a02598","project":"openstack/neutron-lib","branch":"stable/wallaby","topic":"bug/1987378","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-04 13:01:06.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-04-04 16:09:28.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-04 17:45:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16ec53356ec3c5daa1dfa192a1cfa01816a02598","subject":"Skip test_qos_dscp_create_and_update API test in stable/wallaby","status":"MERGED","created":"2023-04-04 10:53:08.000000000","updated":"2023-04-04 20:41:16.000000000","submitted":"2023-04-04 17:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"879441-bug/1987378","meta_rev_id":"581dbbabf5aa1d753c310f70fd291f09fc331272","_number":879441,"virtual_id_number":879441,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 17:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-04 13:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-04 16:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-04 16:09:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-04 10:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 11:11:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-04 13:01:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-04-04 20:41:16.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"}],"messages":[{"id":"f60d2512d92114a866dbf41543e7de02213d68a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 10:53:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a77e2b63a21dc412207cbd23319d321d0f06a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:53:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-lib on branch stable/wallaby.  The error was:\n\n  extra keys not allowed @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027neutron-tempest-plugin-api-wallaby\u0027][\u0027tempest_exclude_regex\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-python3-wallaby-jobs\n        - lib-forward-testing-python3\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - neutron-tempest-plugin-api-wallaby:\n              # NOTE(slaweq): In stable/wallaby we need to skip this dscp marking\n              # related test as fix\n              # https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/854369\n              # is not available for this branch\n              tempest_exclude_regex: \"\\\n                (^neutron_tempest_plugin.api.test_qos.QosDscpMarkingRuleTestJSON.test_qos_dscp_create_and_update)\"\n          # This job comes from lib-forward-testing-python3 template,\n          # but it is limited to 2h there and we want to set\n          # for it 3h timeout\n          - tempest-full-py3:\n              timeout: 10800\n      gate:\n        jobs:\n          - neutron-tempest-plugin-api-wallaby:\n              # NOTE(slaweq): In stable/wallaby we need to skip this dscp marking\n              # related test as fix\n              # https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/854369\n              # is not available for this branch\n              tempest_exclude_regex: \"\\\n                (^neutron_tempest_plugin.api.test_qos.QosDscpMarkingRuleTestJSON.test_qos_dscp_create_and_update)\"\n          # This job comes from lib-forward-testing-python3 template,\n          # but it is limited to 2h there and we want to set\n          # for it 3h timeout\n          - tempest-full-py3:\n              timeout: 10800\n\n  in \"openstack/neutron-lib/.zuul.yaml@stable/wallaby\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"c35d8d41caaa471333131fca5aff0d392bfb50ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-04 11:11:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36022d1191b73e50f7481554c5d8125e10b9d375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 12:23:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4602d4a3227f3beb360a381c3ce076be7ad7b765","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 12:28:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74fbf6b3e0a9f0e7fd1c18d2875ab15a647d5067","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-04 13:22:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d0ed113d327bafa22557a8e64c31f6145e87c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 13:42:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee2098b10814513b93e5e1443988b4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02ca02af3304c9c8f7b454e84153a4e : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/286e5101a5364838a6813cd10e6a54d2 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7828607103ff433a83fef0b9b7657638 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6962a2f0a3b4eaf9ea7dbc40a00bfbd : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/867bc8bf074e409e8121915dac6ba147 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb5e50d3b08e48d4952208db73cf1b1d : SUCCESS in 3m 54s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89b8fc4d6f76424c9e70f85e758a8e2b : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/8906d0e1930b4aee9154c53954ae45a0 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":2},{"id":"05fba9b960ed2b9f7cdfae4aae56ad67079a347b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-04 16:09:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11835588eee174175be3353afb11b3bb77f39314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 16:10:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"59cd5dd5151c0befd615d3c76cc968b75650099b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:45:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa6afd4b1339428bb0811125ad9ecf2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47ed5652e904327ab243862639b1951 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1428f54cb64f93ab2dfc6b155c605c : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9551208b4163469c9723ddcf380c6c7f : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffed53307ac94c2fa4e4ce91e3e25af7 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21618ca5b32f44cfa4a2a1d56d168fc2 : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/a3cac73e07ec4a29808f16ae8109219a : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e9745f741dce7b3ee04f7a1ca13a3cc3ea2a2d84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:45:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d0f1f0c9bec238f9c36469151e140436590805e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:48:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e864a01c8384a2abc1b875913e6c225\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb31360fd1b64081b88440723a6d3ff8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2},{"id":"581dbbabf5aa1d753c310f70fd291f09fc331272","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-04-04 20:41:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ed946347c471e11bdc43ee117782e892de169f6","revisions":{"21ef9eccb93a99292941cd4667103ec572839771":{"kind":"REWORK","_number":1,"created":"2023-04-04 10:53:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/879441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/879441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/1"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 10:49:01.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 10:49:01.000000000","tz":120},"subject":"Skip test_qos_dscp_create_and_update API test in stable/wallaby","message":"Skip test_qos_dscp_create_and_update API test in stable/wallaby\n\nFix for this test for branches which don\u0027t supports DSCP mark 44 is\nin neutron-tempest-plugin [1] but it\u0027s not in version used for\nstable/wallaby branch and there\u0027s no way to backport it\n(neutron-tempest-plugin repo is branchless).\nAs stable/wallaby is already in EM phase we can simply skip this one\nfailing test in that branch to unblock the gate.\n\n[1] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/854369\n\nRelated: #1987378\nChange-Id: I16ec53356ec3c5daa1dfa192a1cfa01816a02598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/21ef9eccb93a99292941cd4667103ec572839771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/21ef9eccb93a99292941cd4667103ec572839771"}]},"branch":"refs/heads/stable/wallaby"},"7ed946347c471e11bdc43ee117782e892de169f6":{"kind":"REWORK","_number":2,"created":"2023-04-04 12:23:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/879441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/879441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/879441/2"}}},"commit":{"parents":[{"commit":"acfb2fdc29ab6b22003903b16688cde373f5082a","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 10:49:01.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-04 12:23:18.000000000","tz":0},"subject":"Skip test_qos_dscp_create_and_update API test in stable/wallaby","message":"Skip test_qos_dscp_create_and_update API test in stable/wallaby\n\nFix for this test for branches which don\u0027t supports DSCP mark 44 is\nin neutron-tempest-plugin [1] but it\u0027s not in version used for\nstable/wallaby branch and there\u0027s no way to backport it\n(neutron-tempest-plugin repo is branchless).\nAs stable/wallaby is already in EM phase we can simply skip this one\nfailing test in that branch to unblock the gate.\n\n[1] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/854369\n\nRelated: #1987378\nChange-Id: I16ec53356ec3c5daa1dfa192a1cfa01816a02598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7ed946347c471e11bdc43ee117782e892de169f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7ed946347c471e11bdc43ee117782e892de169f6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
