)]}'
{"id":"openstack%2Fneutron-lib~880103","triplet_id":"openstack%2Fneutron-lib~stable%2Fzed~Ibde6acdb99555a6e43ca253523df7cbe4d150787","project":"openstack/neutron-lib","branch":"stable/zed","topic":"bug/2015987-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-19 09:36:05.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-04-19 09:36:05.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-19 07:44:38.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Ibde6acdb99555a6e43ca253523df7cbe4d150787","subject":"Return properly elevated context by get_admin_context() helper","status":"MERGED","created":"2023-04-12 13:07:42.000000000","updated":"2023-04-19 09:37:05.000000000","submitted":"2023-04-19 09:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880103-bug/2015987-stable/2023.1-stable/zed","cherry_pick_of_change":880102,"cherry_pick_of_patch_set":2,"meta_rev_id":"e4c5c6143030b5eddcd1dc994a72d0e9710194c3","_number":880103,"virtual_id_number":880103,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-19 09:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-04-18 12:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-04-19 07:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-04-19 07:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 13:07:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-04-12 14:41:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 14:44:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2023-04-12 14:44:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"},{"updated":"2023-04-12 14:44:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-04-12 15:44:57.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-04-18 12:15:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-04-19 07:44:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d83b2385fd1fedae5374d14072ba127528db2a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 13:07:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dade073fd97df813d7a30d49ba502666a1d84754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:41:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b77f0ea33e4a4f990d1c165b5d8737\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be644264e9cd4050bc57a6f83ee031f8 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/786eae4d49db499991df5afea8a1f6ea : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f613f740204ae1aba8db5adffc326b : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0cc941eeecf436987abe868b5e576b8 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8039d2b3afbf48fb968784fd32bebf2b : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a12bf6d010ab4169b018c6aaecc898a8 : SUCCESS in 3m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bf2c635b7f74ddfa718b958468bf222 : SUCCESS in 1h 26m 14s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/fdfc296da2d146d7a67f41fe86e38ed9 : SUCCESS in 1h 22m 38s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/9876b111bfe04d92a13cb238f1efedeb : SUCCESS in 1h 07m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/293577859aa84db9aa55ba70c165360b : SUCCESS in 1h 16m 09s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/5b8891077a164713ba74257e943d1d17 : SUCCESS in 1h 29m 38s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/7f597d268221432e9dfbc91ec26d6ed0 : SUCCESS in 32m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2381736ce97a49bfa8f4e9e9183bb19b : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fadb5fd4c0648be37501ba4c8956a178c7c4f224","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-04-12 15:44:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"daa87393ee77fd29b852db6ec188bc6af28e254d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-12 18:08:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a43cda27dbb642ffc446132aeebc2593204ce9fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:57:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718f9ea445d0425b9a475dc5e331b6b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb698543eab4b6596d693303e8d5903 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/961feb57bc914c0b847f62e9ef6d9d23 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985e8a0b897841f796e1484dbe2a2a71 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3876e804d1c4746a624dc7c4dffcfcf : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/553dc644418f4b6eac7446bd4a1409fc : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c26ebbf34115402abd44d59bbc4a9be5 : SUCCESS in 3m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55eb97dbd2bc45beb04e281b80ead6c8 : SUCCESS in 1h 34m 01s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/d396692556e64aec856768a1803d7b31 : SUCCESS in 1h 33m 44s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/4d893ffd4488439b99ad5381e0c9a4e9 : SUCCESS in 1h 07m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/2371190ed95f49c9879046c44a2e600c : SUCCESS in 1h 21m 06s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/089ece3f11be4044b2bab27230904934 : SUCCESS in 1h 35m 35s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/9edfc8ef1fa14785b07af44a66a9eef5 : SUCCESS in 30m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b5c5ca5641444f9b373f40e4bd55f3a : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":2},{"id":"371435d5a84e176bcfeb54f734ebfa7d23842b2e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-18 12:15:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2df5c2af77648404332a8b0d561eef060f248dea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-19 07:44:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"829854c4fd0e51df9f765b75cfdb9bb235b073a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 07:45:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89f348d9db38295435632de1a07c0ddfaf4749de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 09:36:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf9a91593f334660acc083269c6099eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c1fa530d1748308edaf14daef0951d : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a65e1d7af794f64b7f87db59f3aac62 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/369cfd82c9924bc7af7ef31a78a9e957 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/902183dd344f471c8adbb3318c8f4ad5 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15f32c1b5bba434893c7a79427471f39 : SUCCESS in 1h 05m 38s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/0fac4187964148f782e685dcdc07e1a0 : SUCCESS in 1h 48m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef99a062b75a45d2b1be002a496243e2 : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1b065dff102e936d0b5a1a4c4e97666945dc1abc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 09:36:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e4c5c6143030b5eddcd1dc994a72d0e9710194c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 09:37:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40b1db5ba544dc5abfbbb8846b2e69f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a79e800892434589468cfa1886b4e3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbfb104f90c66f2fb774cad90067597e1ade13c7","revisions":{"7306bd4fe7895728073640cc79d226bd6e24a2f7":{"kind":"REWORK","_number":1,"created":"2023-04-12 13:07:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/880103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/880103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/1"}}},"commit":{"parents":[{"commit":"9dd3a49edf01baaa1943f382b751950a510c491a","subject":"Merge \"Do not set system scope on context\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9dd3a49edf01baaa1943f382b751950a510c491a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-12 10:02:49.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-04-12 13:07:42.000000000","tz":0},"subject":"Return properly elevated context by get_admin_context() helper","message":"Return properly elevated context by get_admin_context() helper\n\nIn neutron_lib.context module there are two simple helper functions:\nget_admin_context and get_admin_context_without_session.\nBoth returned Context object with is_admin\u003dTrue but without admin role\nset and due to that it wasn\u0027t proper admin context when new secure RBAC\npolicies are used.\nWith this patch both those helper functions returns properly elevated\nContext object with admin role set.\n\nCloses-Bug: #2015987\nChange-Id: Ibde6acdb99555a6e43ca253523df7cbe4d150787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7306bd4fe7895728073640cc79d226bd6e24a2f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7306bd4fe7895728073640cc79d226bd6e24a2f7"}]},"branch":"refs/heads/stable/zed"},"cbfb104f90c66f2fb774cad90067597e1ade13c7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-12 18:08:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/03/880103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/880103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/880103/2"}}},"commit":{"parents":[{"commit":"9dd3a49edf01baaa1943f382b751950a510c491a","subject":"Merge \"Do not set system scope on context\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9dd3a49edf01baaa1943f382b751950a510c491a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-12 10:02:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-12 18:08:26.000000000","tz":0},"subject":"Return properly elevated context by get_admin_context() helper","message":"Return properly elevated context by get_admin_context() helper\n\nIn neutron_lib.context module there are two simple helper functions:\nget_admin_context and get_admin_context_without_session.\nBoth returned Context object with is_admin\u003dTrue but without admin role\nset and due to that it wasn\u0027t proper admin context when new secure RBAC\npolicies are used.\nWith this patch both those helper functions returns properly elevated\nContext object with admin role set.\n\nCloses-Bug: #2015987\nChange-Id: Ibde6acdb99555a6e43ca253523df7cbe4d150787\n(cherry picked from commit 09af59caa93e23447c164b480f6f0f4ec865842c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cbfb104f90c66f2fb774cad90067597e1ade13c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cbfb104f90c66f2fb774cad90067597e1ade13c7"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
