)]}'
{"id":"openstack%2Fneutron-lib~882334","triplet_id":"openstack%2Fneutron-lib~master~I8f1cd2a1375f4c6551791e5cf2427230b687edb8","project":"openstack/neutron-lib","branch":"master","topic":"2023-aa-l3-gw-multihoming","attention_set":{},"removed_from_attention_set":{"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2023-05-05 10:52:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f1cd2a1375f4c6551791e5cf2427230b687edb8","subject":"Remove an unused exception","status":"MERGED","created":"2023-05-04 15:45:41.000000000","updated":"2023-05-05 10:53:45.000000000","submitted":"2023-05-05 10:52:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882334-2023-aa-l3-gw-multihoming","meta_rev_id":"fb987bcbbd329e8be71b72d922c4c9b5eb1c9776","_number":882334,"virtual_id_number":882334,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-05 10:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-05 07:46:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-05 08:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-05-05 08:42:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 17:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-05 07:46:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-05 08:42:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"beb034fa922e97d55bca39391622b4b411a33dcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-04 15:45:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33642c464ddc73e5dfcc8369f159750a0634df6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:58:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b03998a3a75743deb6a5cdbbfbdf1b3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9169b61b2bc746dba8dda2dd5f381fae : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fbf99ec5abc48fdb3a3c5d19611958b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0941814d42c944438d84590a516da0d3 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50d11bf1876b4b80aed9242b6ec6c108 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3d2fe36a34e4a5998bd254c7e12b538 : SUCCESS in 4m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31d39bbaec09498a83fd489f469a6b92 : SUCCESS in 1h 16m 14s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b6b96b1a086e427cbd701e9e4e9d95b3 : SUCCESS in 1h 52m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3343b983b4d34d10a7f6ff1c7fb54ac7 : SUCCESS in 1h 56m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d03b8a9e63924ce7b8618a848e86a377 : SUCCESS in 2h 09m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/822b9d54ee744319984722575c053e9e : SUCCESS in 1h 46m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f5ceee5745c410d8d2fee7c9d32c581 : SUCCESS in 1h 57m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ef87f7a2a8c483d84787e895a757f1a : SUCCESS in 49m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/415c69a6634a4921a6573f0989c0328e : FAILURE in 1h 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"866256e64e30bc5d7ea13ab4cd279a5bac172065","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-04 22:47:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f2001e07d4ee37f5c0902b470a9d77a430e5e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 00:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed83f0f05c20465383729123818cb999\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a9266bd3464fc691f555983d2c9b19 : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc275cc4fd8145e39ea8136de5a637f1 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f8c5567ec24e86aa3198ad157c57cc : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b8a4ebe70d54bb98084283e964740bb : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c808e2abd9924153b92b883aca27c8c0 : SUCCESS in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33aab8f2a4c74df38570634c72e61b79 : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1550d50e8635485697f8547af655495e : SUCCESS in 1h 56m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/edca56a8e8b941c489fe28d7af6387c0 : SUCCESS in 1h 41m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e9060e9915645408350ef45b2b207da : SUCCESS in 1h 33m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a4198b375c824265aa24aa957ff6d21a : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d9d62e1e530746f0b501e0a1b2d94c73 : SUCCESS in 2h 04m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/69cd44057d71429bbf057f94114705f5 : SUCCESS in 33m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4487a8f3824b402799fb37c14f893bf5 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e7e43c54e160bdf7d7f8a23b8ab6ede184487400","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-05 07:46:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6356c062dddac8fd9347cde708d559faefbbe6f2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-05 08:15:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"702294eeab7b7b66015b400bf9671805e2343b87","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-05 08:42:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71354c33f042c43fd4b627de658f712584754899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 08:42:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66365a0160e239bdfb7e23b11c60935941329952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:52:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51dd8f91a4724f3f96920186f87a5e5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1879796531ca418db741709c92e4998b : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff22a222150481483bf86c9f8ae18f5 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa65b5025a342038fad8437d9497ea0 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4111a521cac48e0bc5f26dfc547ed5e : SUCCESS in 3m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b0a1560fe144dca836c68310c050765 : SUCCESS in 1h 50m 37s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dff40716e9954e1f9214142ec36f93c6 : SUCCESS in 2h 07m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dfeab8522a1049e2879348fd49273076 : SUCCESS in 1h 51m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8a26f7fd5e704ee3a9e4c6ee854b9ea4 : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0875cf9ba13d449eb0d1a361d3c82da6 : SUCCESS in 1h 52m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bdd13b005d82435eabe4fd4613651b06 : SUCCESS in 1h 07m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc0d8b7035d94fd1b0743a200865c875 : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fd095abd6f1011f5cf4085c900b6189155218310","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:52:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fb987bcbbd329e8be71b72d922c4c9b5eb1c9776","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:53:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b011ad9f5a8d4155b6c8691a9a7b3d56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b709255855948129b3f6e85df0aef30 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"010c790a0e78052ce4950d06cac09ddc00ee3faa","revisions":{"010c790a0e78052ce4950d06cac09ddc00ee3faa":{"kind":"REWORK","_number":1,"created":"2023-05-04 15:45:41.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/34/882334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/34/882334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/34/882334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/34/882334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/34/882334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/34/882334/1"}}},"commit":{"parents":[{"commit":"96c11047fb10b01b6e0e984b806e759e999cbf50","subject":"Merge \"New api-def: allowedaddresspairs-atomic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96c11047fb10b01b6e0e984b806e759e999cbf50"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-04 15:35:43.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-04 15:45:28.000000000","tz":180},"subject":"Remove an unused exception","message":"Remove an unused exception\n\ne52a9372f76aa104543ab04cf1723e58da17d379 added an exception which we\nlost the need for in one of the iterations of\nhttps://review.opendev.org/c/openstack/neutron/+/873593\n\nChange-Id: I8f1cd2a1375f4c6551791e5cf2427230b687edb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/010c790a0e78052ce4950d06cac09ddc00ee3faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/010c790a0e78052ce4950d06cac09ddc00ee3faa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
