)]}'
{"id":"openstack%2Fneutron-lib~883345","triplet_id":"openstack%2Fneutron-lib~master~Ic70202d1b41ea64ffd63dc910b7852fe75421fa9","project":"openstack/neutron-lib","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-17 18:10:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic70202d1b41ea64ffd63dc910b7852fe75421fa9","subject":"Context.elevated() method sets all required roles for context object","status":"MERGED","created":"2023-05-17 10:19:29.000000000","updated":"2023-05-17 18:11:37.000000000","submitted":"2023-05-17 18:10:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883345-secure-rbac","meta_rev_id":"81cba31ea9704a61fd68ba9004e403ac86838abc","_number":883345,"virtual_id_number":883345,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 18:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-17 10:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-17 11:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-05-17 11:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-17 10:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 10:22:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-17 11:01:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-17 12:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3875bd74012a4ed48889e7011d693e8d8d89f641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 10:19:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c1e7349da7bb606ed5f120b0e19c289247860fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 10:19:45.000000000","message":"Topic bug/2019946 removed","accounts_in_message":[],"_revision_number":1},{"id":"2716749a02188a3af729aadc7f70eb0b56343b96","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 10:19:50.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"931f6baddd210ad1ba7cdd63250db691206913a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 10:22:49.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"5d056aafe91f0b38ca04b6ed59ad7cee173653bf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-17 11:01:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c116144a432cd31bfe9a72286c563384d961251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 12:00:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ce0631baccc4b44bd7d64d8300c929f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512435816c1241c99fbc2d918c66e771 : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f2d9fad77df4e63a73f6ac239298fe1 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d996c992a367466f964f397a645dc0e3 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9037133253d942d3845b9d8b8e6dc869 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d41b512d2b8469f88c47355296859f4 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/331c9feedc2443ccbcf9ab6d5ea212d2 : SUCCESS in 3m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f620d3c6c0754772b5575a4eb8479ac2 : SUCCESS in 1h 33m 36s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ddbb11ecf2c147bca77ea90c7cc856a8 : SUCCESS in 1h 31m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6119ad40f4d4acab2bd8a06780dc074 : FAILURE in 17m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8780096a944c4557a053e9d18756f4f4 : FAILURE in 18m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88535143f55a4d92a2cb524de7a8ddef : FAILURE in 17m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/046c18362bdf48a080bb180a2285547c : SUCCESS in 1h 24m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0ac29a28ed942099d0f95462f72a21a : SUCCESS in 31m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5b91cc012a84d37b4c9e62644de4d8d : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"64536d3de7eb737978788c77a02c7b94f0867d2a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 13:32:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6363311144faff59429163a95e08d74bfc5c7f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:11:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0076d851af0b495089f91d8c13ead3ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b103e08c5c4f87a2b3a20572d8b4e9 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de2984cdc953413ba90c3c714ef919ea : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e46280944c4e7e9426f0073db68780 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe534d0002044044bfefcd4d6a9acf1e : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b14ef47b940843c0a94b184382d58b26 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f630d18be5454608af1cbc1c0f80f8d6 : SUCCESS in 3m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c76c5307a4cc49fc9ddd45f509484e1b : SUCCESS in 1h 14m 58s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/606257f324df491fba08c4601c743ea2 : SUCCESS in 2h 36m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3bd732d3af7441ffa1e5d305c0909726 : SUCCESS in 1h 53m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b883f917e50c40b88d57ce5a8565273a : SUCCESS in 2h 05m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/29f80ee8f1204d02a862af684d0567ea : SUCCESS in 1h 51m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27572e33688a4818876a21c357af1239 : SUCCESS in 2h 17m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c0ffdca68a9a488b84b6d49d587244a9 : SUCCESS in 42m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3499cf51bcc846ea95d8dcf33eeb7004 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":1},{"id":"89554bbd94308cf54c49df9806d6ef95a8b48936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:11:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20ecdb20c8084922d8b7b5b11cfcae9ec9a4799b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 18:10:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5fbfb6a037a4f749fa6e6e65c327441\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4880ad08ff14eb28bca82a5241bdabd : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fce26158a44e5a949d5fdf290792f0 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aa4cb5963584604ae9bd39b8066fb92 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfc3cfcbc4024a398f291afd885e1611 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac6f648bbfc74694bfd26409fa679865 : SUCCESS in 3m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b8c206681ed435d9183f1d38fd10e40 : SUCCESS in 1h 31m 15s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/19e6bcfbe3bb41c8813fe1d268379f42 : SUCCESS in 1h 58m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ed995581dc6479baa365cbab75f0563 : SUCCESS in 1h 19m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9063b3e4ee046148468a42c241c4c2d : SUCCESS in 1h 30m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f86b5fb65ac43e0a1ea19742efac6b3 : SUCCESS in 1h 25m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a7a8738d3a0b4d19a96c8e14c0301509 : SUCCESS in 1h 23m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/957d6e7e6a784a48b0772e286223e601 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e4596bb1f577c7c5e2a888feca66fbcacb25e7a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 18:10:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81cba31ea9704a61fd68ba9004e403ac86838abc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 18:11:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5661179879145f4acda7e6bbed1f703\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f4ad04395547179146b508fbe5762f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5ca1ddf420b827e4684dee6a6495475014a91e3","revisions":{"c5ca1ddf420b827e4684dee6a6495475014a91e3":{"kind":"REWORK","_number":1,"created":"2023-05-17 10:19:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/883345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/883345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/883345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/883345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/883345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/883345/1"}}},"commit":{"parents":[{"commit":"ab8a9dd0d9ea072fdae43c011bba32e6370805aa","subject":"Merge \"New ``network-ha`` API definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ab8a9dd0d9ea072fdae43c011bba32e6370805aa"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-17 10:17:17.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-17 10:17:17.000000000","tz":120},"subject":"Context.elevated() method sets all required roles for context object","message":"Context.elevated() method sets all required roles for context object\n\nIf context should be elevated, it should always have \"admin\", \"member\"\nand \"reader\" roles set as admin user always have \"member\" and \"reader\"\nrole as well.\nUsually, when context is created by keystone it is like that but in some\ncases, e.g. when noauth middleware is used instead of keystone it\u0027s not\nlike that and then context from the environment don\u0027t have any role set\nso we should make sure that elevated context have all required roles set\ncorrectly.\n\nCloses-Bug: #2019946\nChange-Id: Ic70202d1b41ea64ffd63dc910b7852fe75421fa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5ca1ddf420b827e4684dee6a6495475014a91e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5ca1ddf420b827e4684dee6a6495475014a91e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
