)]}'
{"id":"openstack%2Fneutron-lib~883399","triplet_id":"openstack%2Fneutron-lib~stable%2Fzed~Ic70202d1b41ea64ffd63dc910b7852fe75421fa9","project":"openstack/neutron-lib","branch":"stable/zed","topic":"secure-rbac-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-25 20:43:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic70202d1b41ea64ffd63dc910b7852fe75421fa9","subject":"Context.elevated() method sets all required roles for context object","status":"MERGED","created":"2023-05-17 22:01:07.000000000","updated":"2023-05-25 20:44:38.000000000","submitted":"2023-05-25 20:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883399-secure-rbac-stable/2023.1-stable/zed","cherry_pick_of_change":883398,"cherry_pick_of_patch_set":1,"meta_rev_id":"ed710189298bd16335e3be298a68935000eef265","_number":883399,"virtual_id_number":883399,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 20:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-25 15:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-23 13:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-05-25 15:10:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 23:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-23 13:42:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-05-25 15:10:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"01b7e1f3682c0e228defa2107d7fd411f1269ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 22:01:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c2589cc7b9df9ddfc6bbaec2fece68d234c9ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 23:45:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2c7a64994f410c85c77d9d12be1a95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c13b180e7f4c20b76e2643546512c3 : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75aecf6334fc4051acff0c39c25996f0 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a942defe5053471cbdc9e7f27740baaf : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccf3f166cbaa4e6282a77a38d476b616 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dfd218f4c9e4fe986b30acc1f689c22 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5fbbc09f4c6433d91a8a936ae1b6d1f : SUCCESS in 4m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82adfd87d3c646beba594dcae32b8024 : SUCCESS in 1h 25m 04s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/3d701e9d07c541b48e9734d646f64a36 : SUCCESS in 1h 43m 33s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/7184b979c94b4cd8a63fa943c18106f1 : SUCCESS in 1h 21m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/0023838e539a4677998f917a98d901d4 : SUCCESS in 1h 38m 21s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/9e8ee6222fdd48bba68f14d694046e8c : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/692539e0a2f24b99bfb7667470d387d5 : SUCCESS in 30m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ae7ee7ff2eb488e8d684aa792348f1a : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8068991f628bcbca71f9c830ba89ffc5e8d0156b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-05-23 13:42:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"170d5d141e70cc5c9be24827000b56321dffb5f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-25 15:10:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"412a7bd94075975242e2b429c29bab76e8b74b06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:11:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15835559d7edd23a4bb6f51456a8d62c33296d0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 16:53:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d40a5057ae124138a9d7830154b00667\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9bc23db9c94832bfcadaa9f3db11c2 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b641450134844446a79d2fc931f6d3d4 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18d0e0aae3cc4c3788b3f279ffec2605 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34cf449d926a418eb8b540139a5ce953 : SUCCESS in 4m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18350a3d98604f73b456ef8939f751e9 : SUCCESS in 1h 10m 38s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6a259f1ede034dd6a32e75fc7bce9a53 : SUCCESS in 1h 37m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2135dc1de2b84f1292b95a13d05e02ea : FAILURE in 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7ec5ea7d9d24acec7e0197307bf70b6ad2e5b4b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-25 17:30:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4103082442119dba3ee4cc3d3db05ea5a4deb476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 19:07:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60dbbdb94b67443980aa47acfa5b9f58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a3d621a0454873b89b95d1bebd8134 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e24721158fe54ea18263826a0d507045 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b539760ff28a428e977f4e752a050949 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fba6cfb3d82c498ca4c08a988bafabe8 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32b4a8c53dc64b5792b4961590f006ce : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5201feec2e79447eb9707d59ca7db6c3 : SUCCESS in 3m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea755eaecf464ccd8b3db31f00c21e7b : SUCCESS in 1h 05m 53s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/a44c65aa74b84fd8bfde5c1276737c9e : SUCCESS in 1h 34m 57s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/cef7c5a846514a618dea070205b60cb6 : SUCCESS in 1h 33m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/00e76b90468e40e394b916834f9f9a65 : SUCCESS in 1h 35m 44s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ea66729f909546428e3a9e5210f7e50c : SUCCESS in 1h 35m 16s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/8f3d8b0855de484a98f85a2eb8196749 : SUCCESS in 41m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f89a02788064484a8bfd584dc224acaa : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":1},{"id":"493535701e41aade78fbf7ac0b3ebcad1e2e98dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 19:08:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2745da6d8ffb7395ce7d6944105a66dd717d2a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 20:43:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ce109990bb465a99fa45013fa3f29e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77eb3c6700b4e329172a6e2cc7c12f4 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea7d45167184aae922ece9f2f0e55b7 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e975837ac6b43359221f11e15c8bc76 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afb230fd60694293adcc9fcf062a2d17 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89aeb5f9d9b64b6b808e428c6b282afd : SUCCESS in 1h 09m 46s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/258c10167e3f48dca1ec9e6a52005352 : SUCCESS in 1h 34m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b581087ee164f62a4225c655e75ad46 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7e0c8d70ce2e812fd37d1b45bf6ea43a848e537e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 20:43:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ed710189298bd16335e3be298a68935000eef265","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 20:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbefe2eae39e4dc48b6783a88afe0695\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95319ee77e21489bace4c85a5dc54370 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00d4c31942d393942050f912ced14e3c85e4b7d2","revisions":{"00d4c31942d393942050f912ced14e3c85e4b7d2":{"kind":"REWORK","_number":1,"created":"2023-05-17 22:01:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/99/883399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/99/883399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/883399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/883399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/883399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/99/883399/1"}}},"commit":{"parents":[{"commit":"cbfb104f90c66f2fb774cad90067597e1ade13c7","subject":"Return properly elevated context by get_admin_context() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cbfb104f90c66f2fb774cad90067597e1ade13c7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-17 10:17:17.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-17 22:01:07.000000000","tz":0},"subject":"Context.elevated() method sets all required roles for context object","message":"Context.elevated() method sets all required roles for context object\n\nIf context should be elevated, it should always have \"admin\", \"member\"\nand \"reader\" roles set as admin user always have \"member\" and \"reader\"\nrole as well.\nUsually, when context is created by keystone it is like that but in some\ncases, e.g. when noauth middleware is used instead of keystone it\u0027s not\nlike that and then context from the environment don\u0027t have any role set\nso we should make sure that elevated context have all required roles set\ncorrectly.\n\nCloses-Bug: #2019946\nChange-Id: Ic70202d1b41ea64ffd63dc910b7852fe75421fa9\n(cherry picked from commit c5ca1ddf420b827e4684dee6a6495475014a91e3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/00d4c31942d393942050f912ced14e3c85e4b7d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/00d4c31942d393942050f912ced14e3c85e4b7d2"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
