)]}'
{"id":"openstack%2Fneutron-lib~885082","triplet_id":"openstack%2Fneutron-lib~stable%2Fyoga~I800e0356714d59ba93ab6252c77be0a82f024055","project":"openstack/neutron-lib","branch":"stable/yoga","topic":"bug/1918145","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-02 12:13:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I800e0356714d59ba93ab6252c77be0a82f024055","subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","status":"MERGED","created":"2023-06-01 17:14:30.000000000","updated":"2023-06-02 12:14:20.000000000","submitted":"2023-06-02 12:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885082-bug/1918145","meta_rev_id":"d3f0e98bb2cc9ebd84e89e34cfb282ff6207c3cf","_number":885082,"virtual_id_number":885082,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-02 12:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-06-01 18:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-06-02 10:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-06-02 07:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-06-02 10:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 18:31:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 18:33:27.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-06-02 07:41:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-06-02 10:29:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7869ecf080d44bf104b3a6f595458f95dce077fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 17:14:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21e96b30a591cb5cc489134959f202df8101263f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 18:31:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29387fd884c44e37a3b602c12e7b319b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e2dfa30667422085ae22d82d4adc1b : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b7c534e12a4105b53cc108534fccd8 : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/627730844e374a1c8011dba0fb3018e6 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7280d7a511644a8a91dea35e4238d82a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a23a2586a277403dafca77841be5bdf9 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db5ad0d91abf45469064ea5a6adddd58 : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cd86d04a3f443e5b146468d64def791 : SUCCESS in 1h 04m 26s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/cf0ea20940134093b4906055f05607ee : SUCCESS in 1h 09m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80ccfa8461084021a6de59f416dac3f0 : SUCCESS in 47m 35s","accounts_in_message":[],"_revision_number":1},{"id":"68be165c034800d9a1104666df5c79890bf110c1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-06-01 18:33:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f340d60c44b11ca11047fad3977515ea20b9ebc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-02 07:41:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"458e6daebc60f9dbfb4f4331e16b67a833b762a7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-02 10:29:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22a7cf122e525c709c0347b6a252f6c74fe8fe7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 10:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b38499d5d8a345f049aba500b0760aa2c0233e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 12:13:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1bf224709814675b362c73d9e682cc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9698d616309f4afaa246324e63791bfb : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bbb726623524951b1f93b5fa5d4c181 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccaa9916ee2c4201ad06037d627f41ce : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2c45a4582b469a923f1b0202b6e713 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47e987887a5c48cf9700e496cc277a27 : SUCCESS in 3m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba1521aa744c4261bbedd866826e8964 : SUCCESS in 1h 17m 53s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/6766e953545340e290fb7235f33ed5f3 : SUCCESS in 51m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f76dedbcc2584e29bf5cd3b7d0d57d6d : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f69f8191ce9f472aaf13e54f737054fe399a2794","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 12:13:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d3f0e98bb2cc9ebd84e89e34cfb282ff6207c3cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 12:14:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba76960db72344508dbf293f7df9b9dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005fc18f575b4c50ac0ad7a178de29db : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"467eafd20a229e24459ad225b458ff4e04e46fd8","revisions":{"467eafd20a229e24459ad225b458ff4e04e46fd8":{"kind":"REWORK","_number":1,"created":"2023-06-01 17:14:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/885082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/82/885082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/885082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/885082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/885082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/82/885082/1"}}},"commit":{"parents":[{"commit":"5398bbdec6cbb3a3ec27732ab06c1e19508977be","subject":"Use neutron-tempest-plugin-api-yoga job in the stable/yoga branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5398bbdec6cbb3a3ec27732ab06c1e19508977be"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-28 15:43:30.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-29 10:28:00.000000000","tz":120},"subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","message":"Add a \"GROUP BY\" clause on queries with RBAC entries\n\nAs reported in the Neutron patch [1], this change introduce a\n\"GROUP BY\" clause on the SQL queries with RBAC entries. With [1],\nall resouces with RBAC entries (\u0027network\u0027, \u0027qospolicy\u0027,\n\u0027securitygroup\u0027, \u0027addressscope\u0027, \u0027subnetpool\u0027, \u0027addressgroup\u0027) will\nload the RBAC entries with \"joined\" strategy.\n\nBecause of the low cardinality of the RBAC query when all the RBAC\nregisters are in one single project, this patch groups the resource\nqueries by the resource ID. That will reduce the results returned by\nthe SQL engine to only the singular registers required.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/884877\n\nRelated-Bug: #1918145\nChange-Id: I800e0356714d59ba93ab6252c77be0a82f024055\n(cherry picked from commit 829e97024c2b73dd67bfd8a04c65f03be556eec8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/467eafd20a229e24459ad225b458ff4e04e46fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/467eafd20a229e24459ad225b458ff4e04e46fd8"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
