)]}'
{"id":"openstack%2Fneutron-lib~885084","triplet_id":"openstack%2Fneutron-lib~stable%2Fwallaby~I800e0356714d59ba93ab6252c77be0a82f024055","project":"openstack/neutron-lib","branch":"stable/wallaby","topic":"bug/1918145","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-02 12:13:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I800e0356714d59ba93ab6252c77be0a82f024055","subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","status":"MERGED","created":"2023-06-01 17:16:41.000000000","updated":"2023-06-02 12:14:19.000000000","submitted":"2023-06-02 12:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885084-bug/1918145","meta_rev_id":"6a0343730f2d1cfe3cb5a040e3966c6730220aeb","_number":885084,"virtual_id_number":885084,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-02 12:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-06-01 20:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-06-02 10:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-06-02 10:28:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 18:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 20:47:17.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-06-02 10:28:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"b700659bcc9d802f95d129803f0ad9cb179693be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 17:16:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe012fc855b2865d9d710f422013abf002c2f217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 18:56:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d56d4da86c4126a7bf1ba0effd5986\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d446e4af5f14292bb2c616f7c1c8be1 : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77fd485592ff46fb834eb7a32ba31c0f : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f40e3f78b24aad9e4fca87b25e8bee : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99903761a3f441908829b7144cfeedc4 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e2ec6cd3e4244799b44966c883f4cc0 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a614312238d34527aa2620a5a697707b : SUCCESS in 3m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b12cbf6a1b8347d2beec238c32083151 : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/3a28bfec60b24ff288b476200dd043dc : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0d56021bdd1ce81571b9c7d228133fe331a7c354","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-06-01 20:47:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e29f29d534ab8690cdf6fd94cc3b56c5ce71dc71","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-02 10:28:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b300a510eeca6b1af23a46cdb271f0b6e660b135","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 10:29:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb5a1afb20e1c5ed88be98f5322d10920927aae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 12:13:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e027a23a89524d2fb378c64f905f8bd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea09d3837c4a4851895d3e34802da11d : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96207e9e0e4246669138753673a909b4 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/764b93bdcd8440459baf20e54ac93d2e : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08df3b54e9634841b3405eb5a474edf4 : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2d831256a1644aba24c99be12fed936 : SUCCESS in 1h 39m 12s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/73dc20e0c5c540298472f04459c72a95 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"eeb7577dcca858883a9dec897dfcb604c47ce9c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 12:13:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a0343730f2d1cfe3cb5a040e3966c6730220aeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 12:14:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1a34287b1245f3adbbaa2260ef3143\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7a29f12e6a4f8f815d35b4fdb07c13 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"619c0fe5345af66061730ce763a1cd45c353f37c","revisions":{"619c0fe5345af66061730ce763a1cd45c353f37c":{"kind":"REWORK","_number":1,"created":"2023-06-01 17:16:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/885084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/885084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/885084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/885084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/885084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/885084/1"}}},"commit":{"parents":[{"commit":"a2d0842e157293e3934adedfbaae4984894c5508","subject":"Merge \"Fix pep8 errors with pytlint v2.16.0\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a2d0842e157293e3934adedfbaae4984894c5508"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-28 15:43:30.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-29 10:30:11.000000000","tz":120},"subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","message":"Add a \"GROUP BY\" clause on queries with RBAC entries\n\nAs reported in the Neutron patch [1], this change introduce a\n\"GROUP BY\" clause on the SQL queries with RBAC entries. With [1],\nall resouces with RBAC entries (\u0027network\u0027, \u0027qospolicy\u0027,\n\u0027securitygroup\u0027, \u0027addressscope\u0027, \u0027subnetpool\u0027, \u0027addressgroup\u0027) will\nload the RBAC entries with \"joined\" strategy.\n\nBecause of the low cardinality of the RBAC query when all the RBAC\nregisters are in one single project, this patch groups the resource\nqueries by the resource ID. That will reduce the results returned by\nthe SQL engine to only the singular registers required.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/884877\n\nConflicts:\n    neutron_lib/db/model_query.py\n\nRelated-Bug: #1918145\nChange-Id: I800e0356714d59ba93ab6252c77be0a82f024055\n(cherry picked from commit 829e97024c2b73dd67bfd8a04c65f03be556eec8)\n(cherry picked from commit 4a952047823c3c330dad5475d67f77c7ee8d6419)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/619c0fe5345af66061730ce763a1cd45c353f37c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/619c0fe5345af66061730ce763a1cd45c353f37c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
