)]}'
{"id":"openstack%2Fneutron-lib~886589","triplet_id":"openstack%2Fneutron-lib~master~Ie74934754598292b125d2be7edb4bbcbb898a230","project":"openstack/neutron-lib","branch":"master","topic":"bug/2024044","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-07 16:48:20.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-07-07 20:48:22.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-07-06 09:01:27.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-07-07 14:23:17.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-07-07 20:48:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie74934754598292b125d2be7edb4bbcbb898a230","subject":"Removed ``HasProjectPrimaryKeyIndex`` class","status":"MERGED","created":"2023-06-21 10:06:36.000000000","updated":"2023-07-07 20:49:17.000000000","submitted":"2023-07-07 20:48:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886589-bug/2024044","meta_rev_id":"44f467e619153f90f577052ed763ab2ecbefed17","_number":886589,"virtual_id_number":886589,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-07 20:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-07-06 09:01:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-07-07 14:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-07-07 14:23:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 10:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 13:44:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-07-04 16:37:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-07-04 16:37:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-07-04 16:37:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"515bf151d0ebf4666b9c53e0613722846eb8ff2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-21 10:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cc68c26c7eb58a122a6c20fc51b3d40314cb48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:53:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11b82b55635445b790ffb02f65ae3a54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e1a56217cb413eb5258f52648545f3 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5e0741971164f51b6154a807c862740 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0561ff690b4fd8924bbf2fab087a4a : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a0721bcd5844cdb8491138c0fd3469c : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae1562c7bc684c029bdf2f2ad134e96b : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4ec42ca4b6249fd97878c2e050d4e9b : SUCCESS in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c42335aa15aa43648156ba803e96a3ca : FAILURE in 27m 00s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ce60b3bd75f548daaddf07486e6a1d73 : FAILURE in 11m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dc3971c1d6294266b3dec4a6239637dd : FAILURE in 13m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab030196b4874631a73909d7943f7050 : FAILURE in 14m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cf5b8534e2ce46bcadfc9779480707d1 : FAILURE in 13m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/221e23883e81491889c358cdefd55f50 : FAILURE in 12m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ffe71ea4f1ef425cae3f2e59e5b76189 : FAILURE in 15m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3b791a65a244d1ba365c539ad615d3b : FAILURE in 23m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f5946c6c3fdf7b6e7802f258b2d05c2c31196354","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-22 10:29:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"157790e031cba2a9ff5ccfa3434bec5afb962215","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-22 10:31:19.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9f8ea972301de1486e278b7174caae8e9c1cfd73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-03 07:45:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a11effad515ac4c8587bc2717d9d50af567a4d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 09:26:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5166a5ba1f354eb1b5238782572fad3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ed8476516a420682aaabcf070e02ce : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b00bbe30f5464844ae4e72456aad6b77 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b5b2f1736f40ffa7bd590349f3c5c1 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/288102bc6a05441897ee9d24f07d686e : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5ba4b4c0aeb436283dcaf8feac06bcd : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/038aa4c40e8241dda436f737216a46fc : SUCCESS in 4m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7960b2f9a8324472ba376645a81ae845 : SUCCESS in 1h 39m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b45eac55579a408a9f1dadb17e0c9739 : SUCCESS in 1h 34m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3f3b152cf84a40f5ba2e3c51ea84bed8 : SUCCESS in 1h 34m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1a0cfc4a4ff949aeaa6b609daca21c34 : SUCCESS in 1h 14m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d61ec57cd2fe4bfca8d9b706b1e44a3b : SUCCESS in 1h 36m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d1f65ccb86b438eaed8cb85062fba95 : SUCCESS in 32m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9ad68d0356a47e6b7bb6b9c27873abb : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8a0a570fbadb7e484a1d9a291699934855fbcbd6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-06 09:01:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4b87723f2d007b94fe02afe436b4053a5105231","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-07 14:23:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2888c15764a08b536fb4f0dbab8ff42552d9dbc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 14:23:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fcf66be839d711ce805fcd63b684ce5edc70b885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 16:45:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c6396021a564de5977d01fb8ec9419d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c7755481344adf8d706aef52c6f05d : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264468d5057244e8b84d07cb2390639f : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e3d7ce9e3e24e30a73c51534f2491a8 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71abee4dbc174ef7bd3362a2d978b349 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41491b901e5041049b6067baca284125 : SUCCESS in 4m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee21368f8ba24f31b254ef8dadec1450 : SUCCESS in 2h 00m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b61d0736cc204a7989200a16de3e6e6c : SUCCESS in 1h 14m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8c9285f5909d43e5afa0c80864e7b262 : FAILURE in 2h 19m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8339356020d346daae3459ef5f462c7f : SUCCESS in 2h 02m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/339d418437f342de8b862e0d203fb4ef : SUCCESS in 1h 50m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca03309b9edc4731b5abc1068780ea64 : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":2},{"id":"525c04d83b01af6ed33b37a033a50c135e23668e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-07 16:48:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06c693cef95f4bdb6318694c05dca4a973f24ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 18:52:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15f754dfb3d34885942b4ed54864b547\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd84f595089142ebb24fddf49be14a93 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59460993e9cf415fa771b5bdcea4427c : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5d01963b6c46c5bb131a31ff5462aa : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6be47bb85f740d2ad068a231163eed7 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96709eca414448bda81f6afaf166f170 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5e2bbff984b44d78a5da6306902cebe : SUCCESS in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/431e1d80fc184e529cc7bddea0fe5e47 : SUCCESS in 1h 39m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e4a5e5310e34deea3113dd07610ebd7 : SUCCESS in 1h 14m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e194a398df804847ae2c0427a1c5ecd5 : SUCCESS in 1h 24m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e267bf62a0fe48c0bbc8d005c2b21370 : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/36fdc3494dc54b06b6c262f6b0344752 : SUCCESS in 2h 01m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b32a216d10146faa9895c1ed7a1b310 : SUCCESS in 33m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6fb078aaf8c4a05ab28b2347959e659 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e4ec142ebbcfb26eb8e8824cf28d3ecc907a0f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 18:52:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e9825a6ef902326a5a9b5756e9ab50bc23fed27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 20:48:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02d3925382234684a4a7b64449abdcb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c345e93b4b7b4b36bc554b43e314be47 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4841967bcd2f4d32a26c2b5ef8c3aff5 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f421761dd5e942b182d0aa43d712b93d : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f84a82500944ccb8de95a54448ce80 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8e9b7371b0a44118002062226305c74 : SUCCESS in 4m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac4ef2f2a83647a591e3c2be3b9c3868 : SUCCESS in 1h 50m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9719fa82430c43cf8197cd9ca305fd95 : SUCCESS in 1h 32m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa6a377cea8f4b269882284ac1ebdadf : SUCCESS in 1h 44m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c452b2baf2d4e93b1b83d412d1f4d7a : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9cbd335a2ada41929439e6a0084338ae : SUCCESS in 1h 48m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f89cfebb1504c5991661c8c1dd506d8 : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":2},{"id":"86024f5fe1c69fb48a5081223dd149df0ba2a916","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 20:48:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"44f467e619153f90f577052ed763ab2ecbefed17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 20:49:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b365c9cf13433fb8d4617d36ee5454\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e578cb83a6064cb7a47ef0401ec5f272 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"673e48a1890c721654ce0de9cd9e0897c791bd6a","revisions":{"153f15c51627f37ace71cc62234d99063706ae85":{"kind":"REWORK","_number":1,"created":"2023-06-21 10:06:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/886589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/89/886589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/1"}}},"commit":{"parents":[{"commit":"95b6a4bb47347bbd442ac686a805e6bb5c727d56","subject":"Merge \"Add new SG rule ext. ``security-groups-rules-belongs-to-default-sg``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/95b6a4bb47347bbd442ac686a805e6bb5c727d56"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-20 17:11:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-20 17:13:34.000000000","tz":0},"subject":"Removed ``HasProjectPrimaryKeyIndex`` class","message":"Removed ``HasProjectPrimaryKeyIndex`` class\n\nA column that is primary key creates an index by default. There is no\nneed to create another one by passing index\u003dTrue.\n\nRelated-Bug: #2024044\nChange-Id: Ie74934754598292b125d2be7edb4bbcbb898a230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/153f15c51627f37ace71cc62234d99063706ae85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/153f15c51627f37ace71cc62234d99063706ae85"}]},"branch":"refs/heads/master"},"673e48a1890c721654ce0de9cd9e0897c791bd6a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-03 07:45:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/886589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/89/886589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/89/886589/2"}}},"commit":{"parents":[{"commit":"d5acebbe7bf8f987cd06e93b1740a76c50e78af4","subject":"Merge \"Add port hardware offload extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d5acebbe7bf8f987cd06e93b1740a76c50e78af4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-20 17:11:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-07-03 07:45:13.000000000","tz":0},"subject":"Removed ``HasProjectPrimaryKeyIndex`` class","message":"Removed ``HasProjectPrimaryKeyIndex`` class\n\nA column that is primary key creates an index by default. There is no\nneed to create another one by passing index\u003dTrue.\n\nRelated-Bug: #2024044\nChange-Id: Ie74934754598292b125d2be7edb4bbcbb898a230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/673e48a1890c721654ce0de9cd9e0897c791bd6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/673e48a1890c721654ce0de9cd9e0897c791bd6a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
