)]}'
{"id":"openstack%2Fneutron-lib~886945","triplet_id":"openstack%2Fneutron-lib~master~Ib93e6330e4f0e300d96cfe5f7f44aeb82ced917a","project":"openstack/neutron-lib","branch":"master","topic":"bug/2024674","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-26 12:34:37.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-06-26 08:35:06.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-26 12:34:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib93e6330e4f0e300d96cfe5f7f44aeb82ced917a","subject":"Disable pool recycle in StaticSqlFixture","status":"ABANDONED","created":"2023-06-26 06:19:49.000000000","updated":"2023-06-26 12:34:37.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5ec69dab042420de7b8620bc1b5f7f3a37a2fd0","_number":886945,"virtual_id_number":886945,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-26 08:19:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2023-06-26 07:59:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2023-06-26 08:35:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 07:59:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-06-26 07:59:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-06-26 08:19:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-26 08:37:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"901f8b585d98a251c80242e4700e52f5b7f97197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-06-26 06:19:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9688e7344df66fb063026a44466f40c9c81b55da","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-26 07:59:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b441868827aca83e498466ffb5897165ffbea43f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-26 07:59:12.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"6d4373e78b39b1d7b284616caa7288c8fca314e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:19:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2226cbfdccfa4614845c926c2039da36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5576d8f90364fd68602f6f012422806 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c9690427fdc4a33bd6e6ba4ee43a045 : FAILURE in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d260ef8fd640c2b6a9732dc86a9d9f : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e5b3f366564bc2a739a5cfc905c7e0 : FAILURE in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ecf2a1b12d346f28c1f3f60c479a2a4 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/116de69920304510972d88ec77d7492f : SUCCESS in 3m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90e10d2ea3cc4ecd9224ec68e47cfff4 : SUCCESS in 1h 27m 45s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b413d0122237430293dd183ed3c220c0 : SUCCESS in 1h 51m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7595863376b49eb99a1ca795659824f : SUCCESS in 1h 41m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/110fe50f5cac484f8d2f6f32192b0ac4 : SUCCESS in 1h 22m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/63d78178757c49e1a42912e25d201368 : SUCCESS in 1h 39m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8f375d046d94597b8a05ba8edbb4644 : SUCCESS in 1h 57m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aed8b71901a14b598da04cb44dfbf4f0 : SUCCESS in 42m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6737f7943c6e48c8af4641f90d08fc87 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"09d67b72337ba4c91562d5fc0325f30c63298111","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-06-26 08:35:06.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5ec69dab042420de7b8620bc1b5f7f3a37a2fd0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-06-26 12:34:37.000000000","message":"Abandoned\n\nPushed https://review.opendev.org/c/openstack/neutron/+/886982 in neutron instead with config override","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01f77387a7fbe7ebc1ad66e6f2fff808143d0d52","revisions":{"01f77387a7fbe7ebc1ad66e6f2fff808143d0d52":{"kind":"REWORK","_number":1,"created":"2023-06-26 06:19:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/45/886945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/886945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/886945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/886945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/886945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/886945/1"}}},"commit":{"parents":[{"commit":"8c46dd4b6734cd993bbf937c08d1059d88a9906c","subject":"Merge \"Add subnet onboard API ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c46dd4b6734cd993bbf937c08d1059d88a9906c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-06-26 06:09:03.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-06-26 06:18:17.000000000","tz":330},"subject":"Disable pool recycle in StaticSqlFixture","message":"Disable pool recycle in StaticSqlFixture\n\nThe default for connection_recycle_time is\none hour. If any test using StaticSqlFixture\nruns after 1 hour it fails as connection get\u0027s\nrecycled.\n\nWith sqlite memory db if there is a connection\ndisconnect or reconnect db get\u0027s wiped off.\nThis patch disables the pool recycle so tests\ncan run fine even in slow environments.\n\nCloses-Bug: #2024674\nChange-Id: Ib93e6330e4f0e300d96cfe5f7f44aeb82ced917a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01f77387a7fbe7ebc1ad66e6f2fff808143d0d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01f77387a7fbe7ebc1ad66e6f2fff808143d0d52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
