)]}'
{"id":"openstack%2Fneutron-lib~887013","triplet_id":"openstack%2Fneutron-lib~master~I1a85d1af7b4f50a5319c64e47c52ccdaa345e362","project":"openstack/neutron-lib","branch":"master","topic":"router-flavor-ovn","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-11 18:43:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-07-17 14:55:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a85d1af7b4f50a5319c64e47c52ccdaa345e362","subject":"Add new callbacks event for resouce status update","status":"MERGED","created":"2023-06-26 23:43:32.000000000","updated":"2023-07-17 14:56:32.000000000","submitted":"2023-07-17 14:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887013-router-flavor-ovn","meta_rev_id":"de8170cc60ad965af6c3474d3dfd393bcfd5be41","_number":887013,"virtual_id_number":887013,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 14:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-07-11 18:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-07-17 12:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-12 07:30:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-07-17 12:54:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 01:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-27 10:29:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-12 07:30:01.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-07-17 12:54:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ad4ab84365be6c2413d6eb69c690d1d32bca6190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-06-26 23:43:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afd569c225bcf8964f79d658f085245eb3122a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 01:30:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c45279f97764a2397255259c1d19726\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92748662b5842548c48b7f252f871ab : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b04c2a00f7f0448d832e524d85be11e7 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0538c43217e43c991f93e41b7af9b5c : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a5c8220aadf4e31aa5b2b93aca0c5e9 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db43462cc559416b8806414bc7bf4b75 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80985f61e7c2457193cb20f8e6f5dc14 : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7edf9a89eca24f748d04257c6ce4bad4 : SUCCESS in 1h 39m 42s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f89f360213b54b298810d3a450b76b2e : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7f1dc5e056c4b3fb627267eac4131b1 : SUCCESS in 1h 29m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ffb058b055ff4f0db3655a3cef61b7eb : SUCCESS in 1h 28m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb7078b5aa3345d293aeec202a5adb7f : SUCCESS in 1h 22m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce62e92c5ecd43cdaade96459bfaa9bd : SUCCESS in 1h 35m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/117f067039404132a7ef2727f5c2bf90 : SUCCESS in 37m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b41c4b425a14c1f846989fe14402f3e : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f6dbe97903ca2935ab793266e7deab8692f50489","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-27 10:29:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d934f198d4fac7abf778a241a3c0f5da6b11f685","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-06-27 18:42:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b49da1c8ae1a4a1a3152ae3dfc43a2aa3ced8de3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-11 14:53:36.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2fcf7d5539f276a8765f51e675344f267742c788","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-11 14:54:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"595b7f014ec241bbd908a70c9e9661f3dde286e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-07-11 17:43:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6743df151b5c979a71ff6b7bf61bd4c3141a8dba","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-07-11 17:44:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3b6097198d4ce5f1dbbfb9b5b02df7309bdc02a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-11 18:43:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6da404b697ee18b03a9f4d5a7c7a6229463e78bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 19:41:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d2f18d189c349eda1851f0d289e02d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc7ef5e0d234f968960500268b0f455 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7c656fc64eb44bc8fd96df4264cbaae : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4581cb12951c474cbfa9efa9be26e4af : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0bccc1987324a6f8afa7e61c20d8558 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c355d96ee2994061a2310ba6a866122c : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f9705c28fbf41a68688b1468fd236f4 : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c59b694fc374669bcd4b6517f877cf4 : SUCCESS in 1h 13m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5c60005efe54731b66cd8b3383fe1bc : SUCCESS in 2m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b73544cb21fb4cef9a8f852e477408d0 : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/432328b277e1482da7bdb54f7abb38c6 : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/70f2c273d25d45e5975e35320a1d8f03 : SUCCESS in 1h 42m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0e8b7ec8012b4b9cb2af8dfa6c9943ba : SUCCESS in 1h 56m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e9d9d55b1ae348e8a5d3b9d2e2bfa1ad : SUCCESS in 33m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/548a36b98100417ba71978501fa8fed0 : SUCCESS in 1h 12m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4396e2e2c7701bc4bc1998e6f261a939da5ffa1c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-07-12 07:30:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"595c2117428c01310e12202c5054edb45ad1802c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-17 12:54:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"51c007f3871bd6c6f3f3c8a83a9186021733c32d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:55:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"27b600a5812fbd0bbbb6e6effacc4da5caa034b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:55:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/696bc79cc9bc472db37dfbc47f2f6e06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9e02aca8fc4d378f967615d5fe0989 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318e24379cd742439bd6a9afba05e574 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08df9cd0fedf4dc599446768eda41fb1 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/547c616f8d5e4303b17a1ae77d0c3f51 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88bfeec6f98f499eb7cf8eb1b0b8410d : SUCCESS in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8147d02fb8e6454b8444c24006cd0200 : SUCCESS in 1h 53m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14cb4cc1fbb343928da28e5e645ca0e1 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/81d407a246fe4a3197ab9312987179d2 : SUCCESS in 1h 22m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c4d7062b95164cf9bf9a276ac1a37de8 : SUCCESS in 1h 32m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78fca72e4a1d495aa53646fd3002cb98 : SUCCESS in 1h 27m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4526f756f6aa4c99802ed2e0dca69a33 : SUCCESS in 1h 17m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38be4ff6070d43f48bcfe6f7f73c227d : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6c9f156829f9f594cb3f533a056139f06c9277d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:55:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"de8170cc60ad965af6c3474d3dfd393bcfd5be41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:56:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741cf935f56041df9e0383158bba9f7e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e675f07b03d44cd9c0a8808bbb7f980 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f272386519374530ae865324de701e9f : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c958933fa023732200d96e12bb2a817389d6d9ac","revisions":{"3dcf251f06da00a078cbe7ba0db2416b6e7eb6c4":{"kind":"REWORK","_number":1,"created":"2023-06-26 23:43:32.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/13/887013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/13/887013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/1"}}},"commit":{"parents":[{"commit":"93a97d3be0bb9d0b9d126521a5480bcff57adda6","subject":"Merge \"Add \"network\" to the EXT_PARENT_RESOURCE_MAPPING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/93a97d3be0bb9d0b9d126521a5480bcff57adda6"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-06-26 23:39:03.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-06-26 23:39:03.000000000","tz":-300},"subject":"Add new callbacks event for resouce status update","message":"Add new callbacks event for resouce status update\n\nA new event AFTER_STATUS_UPDATE is added, that will be published to\ncallbacks when a resource\u0027s status is updated in the DB. The first use of this\nnew event is to indicate when the L3 DB layer has updated floating IPs\nstatus.\n\nChange-Id: I1a85d1af7b4f50a5319c64e47c52ccdaa345e362\nPartial-Bug: #2020823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3dcf251f06da00a078cbe7ba0db2416b6e7eb6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3dcf251f06da00a078cbe7ba0db2416b6e7eb6c4"}]},"branch":"refs/heads/master"},"c958933fa023732200d96e12bb2a817389d6d9ac":{"kind":"REWORK","_number":2,"created":"2023-07-11 17:43:10.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/13/887013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/13/887013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/13/887013/2"}}},"commit":{"parents":[{"commit":"8e5b6955d21def2b5223b1e1a32130c57b61b2f4","subject":"Merge \"Removed ``HasProjectPrimaryKeyIndex`` class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8e5b6955d21def2b5223b1e1a32130c57b61b2f4"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-06-26 23:39:03.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-07-11 17:42:50.000000000","tz":-300},"subject":"Add new callbacks event for resouce status update","message":"Add new callbacks event for resouce status update\n\nA new event AFTER_STATUS_UPDATE is added, that will be published to\ncallbacks when a resource\u0027s status is updated in the DB. The first use\nof this new event is to indicate when the L3 DB layer has updated\nfloating IPs status.\n\nChange-Id: I1a85d1af7b4f50a5319c64e47c52ccdaa345e362\nPartial-Bug: #2020823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c958933fa023732200d96e12bb2a817389d6d9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c958933fa023732200d96e12bb2a817389d6d9ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
