)]}'
{"id":"openstack%2Fneutron-lib~887659","triplet_id":"openstack%2Fneutron-lib~master~I85d8abf76712b069402cce69bf2c049ac51496c1","project":"openstack/neutron-lib","branch":"master","topic":"bug/2012332","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-10 14:05:56.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2024-06-26 14:44:02.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-06-26 14:44:02.000000000","reason":"Change was abandoned"},"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2023-07-06 02:15:22.000000000","reason":"\u003cGERRIT_ACCOUNT_28056\u003e replied on the change","reason_account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-06-26 14:44:02.000000000","reason":"Change was abandoned"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2024-06-26 14:44:02.000000000","reason":"Change was abandoned"},"34125":{"account":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"last_update":"2024-06-26 14:44:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I85d8abf76712b069402cce69bf2c049ac51496c1","subject":"Remove the dependency of allowedaddresspairs on atomic extensions","status":"ABANDONED","created":"2023-07-05 08:42:07.000000000","updated":"2024-06-26 14:44:02.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7036062534e4f50ccc0a3442140ddd570eaff5ec","_number":887659,"virtual_id_number":887659,"owner":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-05 11:15:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":-1,"date":"2023-07-10 14:05:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-07-06 02:15:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":-1,"date":"2023-07-06 11:43:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-07-07 03:15:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-05 11:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 02:10:32.000000000","updated_by":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"REVIEWER"},{"updated":"2023-07-06 02:12:22.000000000","updated_by":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-07-06 02:12:22.000000000","updated_by":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-06 02:12:22.000000000","updated_by":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-07-06 02:12:22.000000000","updated_by":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-07-06 11:43:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7457ebf1fb4d499536b137c68648a3e6aa1ea696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-07-05 08:42:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecd1aa0e923f1ae362186d42df786e68f79298fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-07-05 08:46:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a01aec49eb6e65a064f55414fc75fe8288b513a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-07-05 08:49:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f7cfcd3a53e14036fea4bea9a0849ceaab2b182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 11:15:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de99ab63d3684b1aaa5afdc3b0ec23f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94587048f6e04a4084fd3cae91e2758c : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a0ae5935f1c4863aea5339604c2b547 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4350b2d5174c47cdb9eb6402e8253f43 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9943c1cc7ab4ca0baf58a351a94dae2 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b967cff281714c5d85a8c6328df3d4a4 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98f175498b9245e28cfbf320ee85a235 : SUCCESS in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53a714e1d1844824883c7172e4ec5057 : SUCCESS in 1h 15m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b711a3cbe142431a92c824d0db19cc96 : SUCCESS in 1h 48m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7a279295c8c453da81c447127eb4c59 : SUCCESS in 1h 53m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4c63d49586c94f9d90c004ffc2193529 : SUCCESS in 2h 07m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3537a38a59a7474d953f7e86a5a3c084 : SUCCESS in 2h 24m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c80697ca5df4a93b2c2ef9ee95b83d2 : SUCCESS in 32m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15518cdd23d6456ba1b2eb64784bce23 : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":3},{"id":"7edce8dd9febd222f0a45e8d71d81aa76b7bdd75","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-07-06 02:10:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f35891d710536597c632e03180b5d89fb527b624","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2023-07-06 02:15:22.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95cbcb3b2051f2fd0a4dd4297883ce1c0c92a352","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-06 11:43:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f2e6f0a465c68ff4c72fc0dfa5f0d7b5b8c2cf3","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-07-07 03:15:43.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"160fd999f633d3a9ff79d80dbb86ad9943f31c7e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-10 14:05:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7036062534e4f50ccc0a3442140ddd570eaff5ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-26 14:44:02.000000000","message":"Abandoned\n\nFeel free to restore this patch and address the comments","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b8c383e7fb43093850d36cae46336142a247cc1a","revisions":{"990b92faaebc9d7533349adbc321dc42ceb9994e":{"kind":"REWORK","_number":1,"created":"2023-07-05 08:42:07.000000000","uploader":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"ref":"refs/changes/59/887659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/887659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/1"}}},"commit":{"parents":[{"commit":"8ccdecc7d1599119e3f020051449b547db89ee3e","subject":"Add is_service_role property to the context class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ccdecc7d1599119e3f020051449b547db89ee3e"}]}],"author":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-07-05 08:35:16.000000000","tz":0},"committer":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-07-05 08:42:02.000000000","tz":0},"subject":"Remove the dependency of allowedaddresspairs on atomic extensions","message":"Remove the dependency of allowedaddresspairs on atomic extensions\n\nThe dependency of allowedaddresspairs.ALIAS on atomic extensions\nwill cause the unit tests of neutron to fail.\n\nPartial-Bug: #2012332\nChange-Id: I85d8abf76712b069402cce69bf2c049ac51496c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/990b92faaebc9d7533349adbc321dc42ceb9994e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/990b92faaebc9d7533349adbc321dc42ceb9994e"}]},"branch":"refs/heads/master"},"b00e9e4a3bc62cd627b9a07e588adae26ce9be6b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-05 08:46:38.000000000","uploader":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"ref":"refs/changes/59/887659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/887659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/2"}}},"commit":{"parents":[{"commit":"8ccdecc7d1599119e3f020051449b547db89ee3e","subject":"Add is_service_role property to the context class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ccdecc7d1599119e3f020051449b547db89ee3e"}]}],"author":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-07-05 08:35:16.000000000","tz":0},"committer":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-07-05 08:45:23.000000000","tz":0},"subject":"Remove the dependency of allowedaddresspairs on atomic extensions","message":"Remove the dependency of allowedaddresspairs on atomic extensions\n\nThe dependency of allowedaddresspairs.ALIAS on atomic extensions\nwill cause the unit tests of neutron [1] to fail.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/880922\n\nPartial-Bug: #2012332\nChange-Id: I85d8abf76712b069402cce69bf2c049ac51496c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b00e9e4a3bc62cd627b9a07e588adae26ce9be6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b00e9e4a3bc62cd627b9a07e588adae26ce9be6b"}]},"branch":"refs/heads/master"},"b8c383e7fb43093850d36cae46336142a247cc1a":{"kind":"REWORK","_number":3,"created":"2023-07-05 08:49:23.000000000","uploader":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"ref":"refs/changes/59/887659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/887659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/887659/3"}}},"commit":{"parents":[{"commit":"8ccdecc7d1599119e3f020051449b547db89ee3e","subject":"Add is_service_role property to the context class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ccdecc7d1599119e3f020051449b547db89ee3e"}]}],"author":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-07-05 08:35:16.000000000","tz":0},"committer":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-07-05 08:49:01.000000000","tz":0},"subject":"Remove the dependency of allowedaddresspairs on atomic extensions","message":"Remove the dependency of allowedaddresspairs on atomic extensions\n\nThe dependency of allowedaddresspairs.ALIAS on atomic extensions\nwill cause the unit tests of neutron [1] to fail.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/880922\n\nPartial-Bug: #2012332\nChange-Id: I85d8abf76712b069402cce69bf2c049ac51496c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8c383e7fb43093850d36cae46336142a247cc1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8c383e7fb43093850d36cae46336142a247cc1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
