)]}'
{"id":"openstack%2Fneutron-lib~887724","triplet_id":"openstack%2Fneutron-lib~master~I7aadac451c59388c3d647f4ec5380e18a72c67be","project":"openstack/neutron-lib","branch":"master","topic":"bug/2013228","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-17 14:55:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7aadac451c59388c3d647f4ec5380e18a72c67be","subject":"Fix the default value for ``hardware_offload_type`` field","status":"MERGED","created":"2023-07-05 15:37:13.000000000","updated":"2023-07-17 14:57:25.000000000","submitted":"2023-07-17 14:55:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887724-bug/2013228","meta_rev_id":"6e9b2bb3174013a19dbf5b859d2e0b6504a25830","_number":887724,"virtual_id_number":887724,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 14:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-07-17 12:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-07-05 18:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-07-17 12:57:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-05 17:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-05 18:12:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-07-17 12:57:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4c06521befdbf320fa28f3d23ccf43cc8bf2d727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-05 15:37:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10eb1df8b76389680d79163d63e9c41939badb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 17:47:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58c437c2a7d74bcbaa8648422171b6d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7d39910b9944049af0fa87b15c869e : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29ad7c7661374839acb2895373391d45 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55be988c6e89462a8d079d887f633126 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c987c03fb8c247fc808b2bb779c20db2 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea2de562c89345658998ff8077670fe1 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52b4515a1cfe4629b171556294526a62 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49ae19424d5446edbf0f7bd4646cfde3 : SUCCESS in 1h 41m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c8b363b29c894629b2864a82eb72ebd0 : SUCCESS in 1h 36m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5586681f081548f8a5c103d7c61b6950 : SUCCESS in 1h 55m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b8235d1a45d24236b4af1deb43a28190 : SUCCESS in 1h 40m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2ad83fe5054e43c19d6d2d42d0355b66 : SUCCESS in 2h 09m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f3ec96bf4f14d82b15284089325b72a : SUCCESS in 31m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1959dd2f5ae3461789c51ba92b298f93 : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7f57d6e78df282eff910c5902cb0b71217c99be1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-05 18:12:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"189a37dbdbff06cb4a91fe8c0825942118fd50c1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-17 12:57:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbbdce7101d64cb33f1a19b298f0b71594d10253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:57:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e76197e37db2da277fb8e3cacd0286a225c5621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:55:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdecf1471c97430fae2ce1f1acbebf72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6255b29c2e534426a21fe456fed21dbf : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91aba611841b43cba742598bcada708b : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b3c45947c847afa861526b5b4ed7ce : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8efd9a19e0a6411091da1c4847d2f72a : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1759747f5e7240dd9eef3bfed9aef281 : SUCCESS in 3m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5647620bba784608945bbf63dcb1a02c : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3702e907e5304bd4a1f335ee5a2f2b71 : SUCCESS in 1h 23m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c1aef6a8a738432a8e83afd4a9962537 : SUCCESS in 1h 32m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b376d4aa75b4f89b3128a07082e2d61 : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df5cce149d5c428c9a5d6cfd2693e7ad : SUCCESS in 1h 25m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae545cbbeb7c44f0b7eadc8f6f4c2c23 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9176632e72da094314967b23d0029a821ecdaefe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:55:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e9b2bb3174013a19dbf5b859d2e0b6504a25830","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:57:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f080596b19aa492f86e628e0b2d81347\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52adc8549c6b4934994335c94b1039b3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61029af216957c4765494481227f432b4a664c8c","revisions":{"61029af216957c4765494481227f432b4a664c8c":{"kind":"REWORK","_number":1,"created":"2023-07-05 15:37:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/24/887724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/887724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/887724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/887724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/887724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/887724/1"}}},"commit":{"parents":[{"commit":"8ccdecc7d1599119e3f020051449b547db89ee3e","subject":"Add is_service_role property to the context class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ccdecc7d1599119e3f020051449b547db89ee3e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-04 09:04:42.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-04 09:07:51.000000000","tz":0},"subject":"Fix the default value for ``hardware_offload_type`` field","message":"Fix the default value for ``hardware_offload_type`` field\n\nIn the \u0027port-hardware-offload-type\u0027 extension, the default value for\n``hardware_offload_type`` should be ATTR_NOT_SPECIFIED, in case this\nvalue is not defined in the API call when creating a new port.\n\nNOTE: this extension is still under development thus it is legit to\nmodify it.\n\nRelated-Bug: #2013228\nChange-Id: I7aadac451c59388c3d647f4ec5380e18a72c67be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/61029af216957c4765494481227f432b4a664c8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/61029af216957c4765494481227f432b4a664c8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
