)]}'
{"id":"openstack%2Fneutron-lib~888487","triplet_id":"openstack%2Fneutron-lib~master~Icf0b4dec9420d342431549589f05c639e59a207c","project":"openstack/neutron-lib","branch":"master","topic":"bug/1991965","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-28 17:01:11.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-07-19 16:51:44.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Icf0b4dec9420d342431549589f05c639e59a207c","subject":"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants","status":"MERGED","created":"2023-07-13 16:14:57.000000000","updated":"2023-07-28 17:02:06.000000000","submitted":"2023-07-28 17:01:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888487-bug/1991965","meta_rev_id":"11f245f56001e9f433d825c455884f8c60cff4db","_number":888487,"virtual_id_number":888487,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-28 17:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-07-27 06:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-07-19 16:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-07-27 06:14:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 18:21:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-14 14:41:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-07-19 16:51:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-07-27 06:14:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c9cd3cf49b39e9d4ecf9ee5d9ac4b6e0fd96d747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-13 16:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b84ce4988ab3bff70b783a8c226d160fe4e039cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:21:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8ae7010b70477ead3e36f45648e49e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0600b4110d4509a7db49850de69379 : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b16a81b9f8c8481482d028bccc1d1fe2 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c61b1168609421eacb2edc299be1db6 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bda4aff6d7a0435a92064174100c59ba : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/668fe934baca404498dc9e0b59ad4789 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c83a11580124730b58b3f9492627da0 : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a6514835c3645608e19310abd1898fd : SUCCESS in 1h 33m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90bd14791f864ed1bee5666348c39082 : SUCCESS in 2m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5f6b7b83ccb342e2a06fddbacc2d99c7 : SUCCESS in 1h 42m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b704aba204414f078d12b14ad067e21a : SUCCESS in 1h 57m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/36ed5979c5ef4f81ba6a5112dd031d53 : SUCCESS in 1h 45m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ebdddccd3a9b4ff6b3b496e5a3ba025d : SUCCESS in 2h 04m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/773dc4837a7545deb88ffb7f8d056b1e : SUCCESS in 46m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02eb96cce6ed400eb4471b9160117552 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0f32718419b9b9053ebd7dc6e770e0c8d92509dc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-14 14:41:34.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"096f33f178216a8e7925de31f52f1f090330f2a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-19 13:29:16.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b73ae3e5ab824f3c7ec88ac6e506c66bf0ffbff4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-19 13:29:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"83ffeaac7d295e927e45d97f00ee5524b7c295f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:32:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc72a2b09946469ea8fefb56b48e0e25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0115a8f43115419385d96fb5aafa8033 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0144ff4cda44919ad8016a9254f03c : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97d7e713cdc498cae74a0d34741cedb : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/513dbcf698274985b7f5a211ae8ffe50 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/198ef73ca2e94d7f977d74757f728f4c : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/899dc4d652b943ac9a00e1f6b4865d4c : SUCCESS in 3m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ebb0c54ed8844eeb7ba0a6be35a6cf2 : SUCCESS in 1h 26m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcdfa04c3d714cfc9ea8b37c7e99c170 : SUCCESS in 3m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0978b8d2455423a8feed6d82fcf7763 : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/840f021c72f6467fb085fa3bace042d4 : SUCCESS in 1h 58m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5628bec031b64940a604405624c137a5 : SUCCESS in 2h 00m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fcc900ab2950458abe6c0750fbbf671a : SUCCESS in 1h 53m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d48b0cba17e4b6dafabd273aabe068f : SUCCESS in 46m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d052510249343d38ab0a79548ce8c43 : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0e006081b82392cad3e2bc2cbb44133016471888","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-19 16:51:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8296a6effb4176efcc2a255eecc7262bf5aa54bd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-27 06:14:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"77533b691ff6f2578565e92acf9d1ee92c28823d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 06:15:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a120beea57350bacb76f7f77e68fd24a7e3eab89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 09:32:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d890746de966470eb94c32cd62169ec8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc0c305af4d4a8081606bba17c10431 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/991b6b4cd43b4192a17a7170b9a25303 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af7817a24f534498af928c139bd55b88 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb06df5c83de491f9e9ba3d43d549392 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0036fdeebe6d422ca5721586ab235e78 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7705a0d8bfa4276afadb6c11cc071b1 : SUCCESS in 1h 38m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/821faac0909a42e88aeca304af3bce8d : SUCCESS in 3m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a9cd0fece09344eebd50f1f6f4e459eb : SUCCESS in 1h 45m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f347ac5f784843b5bad5e6d6fdcbe10a : TIMED_OUT in 3h 08m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6f5124f1c28d4e38ba1ffe006a3a9c5b : SUCCESS in 2h 10m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/efdca96d85624c89bd2d24f0a20025c3 : SUCCESS in 1h 27m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d64a8b4beb3645b2801a47746fb29c8e : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1837c9a4194acae6e1155a3acdd409b649155fee","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-27 09:41:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40bb891499f64d2d9e6dfa11bede498fd4d26df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 11:43:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9442cb05782e4cb3ae893ce6e04ac05c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51bfd7eaac84469986e2e2f70432659 : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f4e662cca294235847b6c50c8b509cc : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a5c68c488f4ae6bce8f8d54f44c207 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/724099b02ad545588bd11090bf55459e : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fe2edc7af04401fa7ef8e2302992d73 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0366d92f30348aa9086285942b32721 : SUCCESS in 3m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7feb6781e62d4ea492d51aeab1148464 : SUCCESS in 1h 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/464ac01012a54d29a7267c7129b6b7c9 : SUCCESS in 3m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7b12e1fe247643e1928ee3ecf5843dec : SUCCESS in 2h 00m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8dfc0b209b48496882401a4cb151809d : SUCCESS in 1h 51m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/08f8c318587c45d59cb751632561d59a : SUCCESS in 1h 31m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bdf8b817cd3743aa9966f1f195d67280 : SUCCESS in 1h 48m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/43be15cf74364a3fab2b405aaffae4cf : SUCCESS in 34m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e200466809d4479592348ab22f33781a : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0708820e175806d3b9c929d47a765c0df348f9ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 11:44:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43e7c56f0e5fca805b39933ccbfa420303b481db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 13:41:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bddfc77393fc472e9591eb275d16c7d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073ceaef683e4d6ba3bf1fb3ed9e349d : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8037403d693f4e4886670dbc78fe328b : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aefdd6853e82459aaa235f432e2976cd : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6b3ebe2156a433b95490bb0f29a968c : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d2e51c59fec4be3affd0c96e9eac165 : SUCCESS in 4m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54668794a86c4eb182b3ab7fed1612ad : SUCCESS in 1h 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0e2cc973988401b8aa8a62b6a97e4ad : SUCCESS in 3m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4e513fe560204377b8dddb9e88f90cc4 : SUCCESS in 1h 30m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b768fdfcb33d48b5aab2f54f02e9b6f1 : SUCCESS in 1h 55m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eae14b455d9c41f8998ee75ffdde0046 : FAILURE in 16m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0bc044a9551647619072b7617da814d4 : SUCCESS in 1h 45m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a64a8c2c6319444587fa5e89417bcd6e : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d9d054372e6da7098547b13621a7791a86e5fedb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-27 14:26:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb52b663da1b04d07185016cc289975a3a9d9339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:21:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009aeebc38124956befa7efa391e6735\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7e68c0b45e46688f74a131732dfc77 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2df5c83edddd4f35b37bbc5b0dfd756b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979d90d10db5498abe66a01027d36be4 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bff3c19ac3a4364a3d3391a219a0620 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/444683ef83574276b073975e8ff99b0e : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78b0dd5eda494c16bafb6b96805854e6 : SUCCESS in 3m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b46696ffdc34ce58f465c65e031bf7b : SUCCESS in 1h 26m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb181b302784edca577ef0a5d470098 : SUCCESS in 2m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5627952a6c4543b09c11ba34d5bf6fd0 : SUCCESS in 1h 37m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/529018fd2d4f4953a99869d113f99806 : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/977f18b33ee54c8e953fbde9247dd138 : SUCCESS in 1h 32m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6c4cc695c35b4d82a5d425cdd58be4da : SUCCESS in 1h 39m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/81acb40dd2ee44b0a7f53aa321fddd10 : SUCCESS in 58m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fd15f145097415ca51addd867ac01a5 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e09b1553cb971ab8007ada723e39077ffa919394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:21:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33650f7912e3bfe7ea079e5f17bd5175c68173ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 18:22:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/560606aad75e4be4a0dbc6f129cd6d9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53148314ef61473f9f1c8826951ef667 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab19e17525614ad9af4601b50070fa63 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26270790672e4f4f91323cbd584a3b37 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f13bce254f7469ab1f9bbe5b861f915 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcef235b77054934babeb1d762e95aac : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c825b1815e5d4f58ae822d2974e70fff : SUCCESS in 1h 40m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f973483f8714f95978da0858137a08c : SUCCESS in 2m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1ffc42e862594760942102283b3214d8 : SUCCESS in 1h 48m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd39a03b7ec84bbabd3db7b2d135ccfd : SUCCESS in 1h 48m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5adf8e40a80e41399ffab65e46bb43c3 : FAILURE in 1h 59m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ad09e062b2014bcda398394c215996c3 : SUCCESS in 1h 45m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc6bac1520d14b388dc2b56de69ca183 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b5dc3e9b37e80422eab9bb53269f2dedd1054288","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-28 06:26:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f074f7da4b2b95d675f155c9d66118e6943f1d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 08:43:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b581643b399f487d8901a1628552a1a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9313d1aaf391418b9267a345e0723b51 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f24aaf9cb4c462797bc51989af84a60 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeaf84c198264580a993017244bc4028 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34964aa70688431fac6f40de79d88718 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1afab636c2a84da5879fd6c4de6dda6a : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b6403bbc677459d890f892ef0d8a514 : SUCCESS in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8483b9d837694b658c28e076d764f515 : SUCCESS in 1h 21m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160a9e0372004a8b8fefafabab9bddad : SUCCESS in 3m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d882570a1b844074b3c515bbcba77b86 : SUCCESS in 1h 42m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/efffce9df6a842e99be0e00d44630c9e : SUCCESS in 2h 14m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ee1eff8c2e1941b393c716c3b1746c4d : SUCCESS in 1h 39m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b26de4ff89df453dab81c4012673e8bb : SUCCESS in 1h 47m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/63868232b7e34cc4a3c5627728109a85 : SUCCESS in 31m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/31848c283a654de4ab132bcbb81fd8cd : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2df9061cca5998ccb74837536493647014812e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 08:43:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f88733ae2bc5ed19d28f05d711ee3ffec668de1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 10:42:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32816c4307894d9ba6ab435ae7ca6251\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fceed669134858af529d97ebcd71cb : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070ced6556c14c3094faeaf963a0b3c9 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/481fb1173d0c4a03b4173dc6221b39b2 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/844a4e0296dc45c08fa5a9447fc96a68 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63a995379ac94d1c998014d5df801cf8 : SUCCESS in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0033e27746d94c8ea162070990eecd38 : SUCCESS in 1h 18m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34cbdf57b9dd4167949a57e5cef4cfd7 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a27999b357d457087314db34c2eb88f : SUCCESS in 1h 18m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd725405f6154d3697085710afc3afb8 : SUCCESS in 1h 36m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cbb469d6e8340f0a43dd3c3e494cc98 : SUCCESS in 1h 28m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b025b8d828c14e00903fc9c4233dc281 : FAILURE in 1h 57m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b810ef35771842118eaa746cf1d93de3 : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":2},{"id":"96b4f57719273fd02dc88a300ec29d169dd17bd0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-28 13:11:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb56363644454f9c0a5a97595b5c92f65c6e2893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 14:49:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/258c323430ab4dc9a91104577e1b25fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb311840edc44489968dd1f4953d3ce : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1fd8f55675047c6a46572e7adf3b16c : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a92bcae2b0e498da3edc2c374efb2ea : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d970e2f218b4a5baed92e0203a3023e : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de4555e14a0b4ad8b14b8a828d18f32d : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c63f8cdc91a3456cabe0034c528651a4 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/418e79622edc4d7b9d0d8102c01897de : SUCCESS in 1h 15m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03e45d4b897d4c28b2fcd0f963513ca0 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ffe8696235df4a249fda441f64426e87 : SUCCESS in 1h 26m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/539f645ce2ce445b845349552e865f16 : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5568b3784cea4c46b82db4b662ef84e2 : SUCCESS in 1h 26m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6570561c375461b828c7ddac635a9ca : SUCCESS in 1h 22m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/350b7568eeec4e9bb0bf1ca295eed827 : SUCCESS in 34m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/397061785f774b2795d91b4046f084c7 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f4489daaef5a6bed43e43925272e9acb619e7e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 14:50:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15b7a9854739d001681446902ab76d90dc53cf86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 17:01:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa16f2ce00d744bea974819caa371fda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99f3ff4c3b34da0ae3a97df1692e33b : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c86f98e59c4f159f41f5abc2e8faf0 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cdb1f14f5144e7ea367a44657cf31c5 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70b3295cba1a4c138c9644f7e3d011f0 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a7f4afea1c04a54a157ba90ac44b8c2 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18484980573643a680d657b0df3d607e : SUCCESS in 2h 02m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/003a5983a8d74a66b36daea97d47449a : SUCCESS in 4m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bdfc1b85d1b9487f9ea861ff19e404c4 : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9db7818cccb4464dab164f5fb6c5cd77 : SUCCESS in 1h 38m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8353a74f703347bc96fde5ed177756fd : SUCCESS in 1h 40m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9968d9669e044a3e828d9509df9f6a07 : SUCCESS in 1h 17m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc9c88f158254bdcb713c9a86104e8e6 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":2},{"id":"99c8be929b758e27d8f79d4125e0342a66b452aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 17:01:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"11f245f56001e9f433d825c455884f8c60cff4db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 17:02:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac43f502b1a44d79c02c7dab0265407\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d75ea0dec24893868b25f2f14ab6d8 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd631b28d5554909b59dea8d19d22e01 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"884f2add0bdaa7b6eefeadfc5724f3324ab405e2","revisions":{"4d1e3796cd32bd1872b00ce5fbcb475dc55bbd30":{"kind":"REWORK","_number":1,"created":"2023-07-13 16:14:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/888487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/888487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/1"}}},"commit":{"parents":[{"commit":"8e5b6955d21def2b5223b1e1a32130c57b61b2f4","subject":"Merge \"Removed ``HasProjectPrimaryKeyIndex`` class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8e5b6955d21def2b5223b1e1a32130c57b61b2f4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-08 04:17:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-08 04:19:30.000000000","tz":0},"subject":"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants","message":"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants\n\nRelated-Bug: #1991965\nChange-Id: Icf0b4dec9420d342431549589f05c639e59a207c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d1e3796cd32bd1872b00ce5fbcb475dc55bbd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d1e3796cd32bd1872b00ce5fbcb475dc55bbd30"}]},"branch":"refs/heads/master"},"884f2add0bdaa7b6eefeadfc5724f3324ab405e2":{"kind":"REWORK","_number":2,"created":"2023-07-19 13:29:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/888487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/888487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/888487/2"}}},"commit":{"parents":[{"commit":"8e5b6955d21def2b5223b1e1a32130c57b61b2f4","subject":"Merge \"Removed ``HasProjectPrimaryKeyIndex`` class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8e5b6955d21def2b5223b1e1a32130c57b61b2f4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-08 04:17:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-07-19 13:29:15.000000000","tz":0},"subject":"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants","message":"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants\n\nRelated-Bug: #1991965\nChange-Id: Icf0b4dec9420d342431549589f05c639e59a207c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/884f2add0bdaa7b6eefeadfc5724f3324ab405e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/884f2add0bdaa7b6eefeadfc5724f3324ab405e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
