)]}'
{"id":"openstack%2Fneutron-lib~890729","triplet_id":"openstack%2Fneutron-lib~master~Ib34ffe161382673a8406c9b28aa439cf504e7cfa","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2023-10-04 07:57:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib34ffe161382673a8406c9b28aa439cf504e7cfa","subject":"Cleanup py27 support","status":"ABANDONED","created":"2023-08-08 07:27:04.000000000","updated":"2023-10-04 07:57:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15d73c32fe903156219eb1cffd7e35adc734a676","_number":890729,"virtual_id_number":890729,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-09 09:26:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 09:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-04 07:57:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"728124b018209f86e4f1b33d2f3ec7838591f042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2023-08-08 07:27:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4e525ff9814fc514a152b8ba83b7f71c2c8d2f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 09:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d574d24dc3d54852b914f844fcb5a478\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/30286de6e4424cec8af6ca4695e1f847 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c09cb214d9c4a3eb51dfd46b4b4dfcf : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fa659e406944ee3abd3a29ff2e683d1 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6eac1ea6034fb9abda35eef30c51b2 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5e632d993d4e92825213269d855677 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c8960fb3293431496fd12c3d7169023 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0848794a79a4b25855c43f9842d2a51 : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/941ab3e48b924f6dad6f1ca916e8831b : FAILURE in 1h 11m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3722ffe09dcf4e91a6b31625137525ee : SUCCESS in 1h 23m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7196c4796d94121a004f41e43d44b9c : SUCCESS in 1h 56m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d65a0707ab847a4b7c12c8fd168c3b7 : SUCCESS in 1h 17m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5296c7362c94bc595483982fe54ad07 : SUCCESS in 1h 29m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ecf44ea3f01a4ec6aabfd2ab9c0ec492 : SUCCESS in 36m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7bdb01379914056859f9454b6fbe894 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"651ece5ace4807d9902be6e0725c07db07820928","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2023-08-09 05:49:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b22cd56b4211a55d8a97d4336cc3638c17130007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2023-08-09 07:26:04.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5f8a7f247d76108fdf65d504f330a223ee7c081f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 09:26:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81849ddbb7004143ae883c789e8738be\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b455cf7064b4926bb825d9da64341e3 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da6504bc0e14dad9ddbdeeffc1d2360 : SUCCESS in 7m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfbbd0371f3543568d538c1a941ff19a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c475ed3a24cc47878c514feb6b85659b : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00b227e2c50644179cc756a5ad69934e : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ada6ac53c4ad4c2296f9b729d50887a0 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7bc8e2cea4449f0b7dbb3fa3fca7487 : SUCCESS in 4m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc9f3fe3601d4ba5a17be43c3660118d : FAILURE in 57m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7eb32d572bb4d49b458908e5361a163 : SUCCESS in 1h 26m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63e5e8d909724e2baa6add60bdc1f006 : SUCCESS in 1h 50m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4259c48d8eb94a4bbfa70d7586f7bc79 : SUCCESS in 1h 37m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d06ded93c8e54371b412a68af0541351 : SUCCESS in 1h 57m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35950c6cbc344153a37686ae8b7e9f47 : SUCCESS in 37m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f24e4a152d0f40b3b61fecb77319f1d4 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":2},{"id":"15d73c32fe903156219eb1cffd7e35adc734a676","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-04 07:57:42.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d86b5419fad9bef8119d425635b5d521fd26dab0","revisions":{"f7cc6e89b8383a6de1f0ae2c4f05ce1352389cae":{"kind":"REWORK","_number":1,"created":"2023-08-08 07:27:04.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/29/890729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/890729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/1"}}},"commit":{"parents":[{"commit":"4f862daccc18532087d7490b7328731b391e02de","subject":"Merge \"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4f862daccc18532087d7490b7328731b391e02de"}]}],"author":{"name":"wangjiaqi07","email":"wangjiaqi07@inspur.com","date":"2023-08-08 07:26:34.000000000","tz":480},"committer":{"name":"wangjiaqi07","email":"wangjiaqi07@inspur.com","date":"2023-08-08 07:26:42.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: Ib34ffe161382673a8406c9b28aa439cf504e7cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7cc6e89b8383a6de1f0ae2c4f05ce1352389cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7cc6e89b8383a6de1f0ae2c4f05ce1352389cae"}]},"branch":"refs/heads/master"},"d86b5419fad9bef8119d425635b5d521fd26dab0":{"kind":"REWORK","_number":2,"created":"2023-08-09 07:26:04.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/29/890729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/890729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/890729/2"}}},"commit":{"parents":[{"commit":"4f862daccc18532087d7490b7328731b391e02de","subject":"Merge \"Add RP_TUNNELLED and TRAIT_NETWORK_TUNNEL constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4f862daccc18532087d7490b7328731b391e02de"}]}],"author":{"name":"wangjiaqi07","email":"wangjiaqi07@inspur.com","date":"2023-08-08 07:26:34.000000000","tz":480},"committer":{"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","date":"2023-08-09 07:26:03.000000000","tz":0},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: Ib34ffe161382673a8406c9b28aa439cf504e7cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d86b5419fad9bef8119d425635b5d521fd26dab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d86b5419fad9bef8119d425635b5d521fd26dab0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
