)]}'
{"id":"openstack%2Fneutron-lib~895940","triplet_id":"openstack%2Fneutron-lib~master~Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865","project":"openstack/neutron-lib","branch":"master","topic":"bug/2036607","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-11-15 23:07:08.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-15 23:07:08.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-07 18:38:33.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-11-15 20:54:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865","subject":"Add the \"cancellable\" flag to the ``CallbacksManager`` events","status":"MERGED","created":"2023-09-20 15:30:54.000000000","updated":"2023-11-15 23:08:12.000000000","submitted":"2023-11-15 23:07:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"895940-bug/2036607","meta_rev_id":"2520052982f09a8e1d138b401fac2ea539a39ddf","_number":895940,"virtual_id_number":895940,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 23:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-11-07 13:57:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-11-07 18:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-08 09:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2023-11-15 20:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-11-15 20:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 17:28:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-09-20 18:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 20:38:00.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2023-09-21 14:49:48.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-09-28 11:08:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-10-11 09:53:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2023-10-11 09:54:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-10-11 16:00:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"},{"updated":"2023-10-17 11:40:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2023-11-07 13:57:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-07 18:38:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-11-08 09:12:44.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"2bff6791525863ce7bde8dd3052402f5b5ab1731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 15:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e02c91f054f372af192e3a7cc4b3a9a0cd55f3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 16:21:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebbf114a2aa27a53bcf88a7aa517bfa2649ceb0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 16:36:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ae89c68465888e1d62c6866c3e954e25041c719","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-20 17:28:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40682835a1abae4e90769988d99c41641f962c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 18:38:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d127302cf294af2bf77f0c6e5d1a30d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6efc7464ec5f4c6a9e40f48ad9d5d8c4 : SUCCESS in 4m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90287781c44b4f6b9d99e79fc4798871 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f624fd16ef84df29d135c506a3e519e : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7024927bc4694def9c0128ed0067f647 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed7dd10525f4c59ba8efa7cf9735061 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc549ed82f534b6796ae958bad5f2c6c : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bd1c54e44f74762be69d5f53b3c9fd8 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3197593059f6495383ed4e4aef2e7811 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40f3a80aa811437f82410c976f00630b : SUCCESS in 4m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7154a467d43b435f81330dc922ed0191 : SUCCESS in 1h 07m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/483e1417f33945d69e3f825e7fd0c530 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5bf99a33d24444ebde6a8eb929ce2ff : SUCCESS in 1h 43m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e495bf5dba446eabcb0d7b05526d2c2 : SUCCESS in 1h 59m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f0eb295092ac498aa249419e7bf3b735 : SUCCESS in 1h 36m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f4013c030b94b91b029c03281270932 : SUCCESS in 1h 30m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/538166281e0743cb976b74290451b7ae : SUCCESS in 36m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37ce23114d814ed5b875cb98c60756bb : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":3},{"id":"2593d46776d9fb665e4ea17a098c87e8a06a118b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-20 20:38:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8899b7ea114dfe3f1426a3d30c7d5585255413e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-21 08:05:11.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6710418c5132c62a198344bf08e214e8fb365ed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-21 08:31:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"acf0d72e5e131bcf2c461acb4c52660de9d94c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:16:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3419d41d8f9b45e497c7c3c403c5c0ef\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dd1cf612c9544d73995250578bd8a447 : SUCCESS in 4m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f90613e5ce84670aab1a937897286b2 : SUCCESS in 1m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a9b2b308bd493bac194051bbcbb6c9 : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acad3dd7608e4344a4d7e05805353c1f : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9b45918647d4dc7840575e9764f6af8 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc254524b4fe4a74823d227c5cf8f3e8 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55629d4f89ec4ea09cb1e5033b1de144 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef08db677b3949c3930d85e29d10686c : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abc56be7bc404167b303ccb990080152 : SUCCESS in 4m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94494d0d6fd84b2aab3606086cd0bc76 : FAILURE in 1h 17m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d4afc9e4da44ae5a672351e2520167e : SUCCESS in 3m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b8ce2eeb010544efa93b4eeff90dfd73 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e6c7a6eda764522a3bd4907739d2bc9 : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dad14f24483e43f5b9ddeba59d547501 : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/976fb5be399f4185ac7396c190013792 : SUCCESS in 1h 44m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/587a943fff3448178864cbea9f40dc45 : SUCCESS in 39m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/565081d91cc04c71930413a629ac4dc5 : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":4},{"id":"6d6a1d89e488898ed196218ee0f5b55ac1010cc5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-21 14:15:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e530b1d5564870c2be7b5d94ca513914c92caad9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-21 14:49:48.000000000","message":"Patch Set 4: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"41ecd2d4e96aa554999e13fde393adae3644a62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:22:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f15ba847044282929e17373ca0646b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/18e1e13d19e2448a960113794c33cfed : SUCCESS in 4m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/405a1ed9321645a89bc66db1d167f33d : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302ad7daf6714f408646717d835c6fde : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1727f21baaa4fa5b0901ea04847cad1 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d874f7b19f9841c5af1f5f53a3f48c8b : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55e8687485dd467485f40e9c0015ac8a : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36a31270a1ea4a33b5c784acee164eaa : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f475b4b4cab4467a8aa9426be064de0 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7b5bb8daef745d7a5845ee03c677f3f : SUCCESS in 3m 50s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31563b6cabea479c816afd195ea6dcd0 : SUCCESS in 1h 34m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a17cf9ac51754f598904c366f23d86d1 : SUCCESS in 2m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b263fddc321647d6b3f690cadaff8796 : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e2c5dab6b3e44f02ab7524127cb6c276 : SUCCESS in 1h 27m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ce5437b4ed73459f83a8441ef6b4cfd7 : SUCCESS in 1h 35m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/36e6c8fbc9f54e8bb95a7ad43cbf5c26 : SUCCESS in 1h 26m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2d8a4007e004992bdc51de327c04afb : SUCCESS in 46m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5df3d63834ab49928da3efd90e0bb9a7 : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":4},{"id":"f399e9403802d0a83dc14b19d9dbbd71c83fbf2a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 08:25:35.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3e04642f6ca4b1905e95bfe11d689bac8b112504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 08:26:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d390c62de9b28e2be41de5c5f47dd5a7b846a33e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 10:05:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac44c082e0ea4838bc73244c0141cc6e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bdb2daf8159447d8b278172735907983 : SUCCESS in 3m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6843a3317b6940b6b81809c6dfe0cbb4 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6620b041491c404daf0d3d5d866ad51f : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b2ca1879b824643810c1bcc9804d435 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef545ed59534438b82e0c52d4f85f5e : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/956ac1c9dafb455bb313d14471b417b9 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87b17684401e474186d2c9089f432ed2 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7115b13ffcbd4127990e2ec400e5f03f : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a6e234538dc4a9fba9d5dcaaaf710b9 : SUCCESS in 4m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b7a2bac7ca64b4ba3221d2d4abb6f30 : FAILURE in 1h 27m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0758c5f3b444491972ace608e05cac6 : SUCCESS in 3m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7769a94b723148b1b9d93f163fa5608d : SUCCESS in 1h 30m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f63649725a9147799bda2ee7daa450b5 : SUCCESS in 1h 38m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93707a3355c0428086cf910c88ef1725 : SUCCESS in 1h 26m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a9df40174614fc1813c969efb4f9043 : SUCCESS in 1h 33m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c01b89d3ddc4ef6bc3db50b6ef5d2b2 : SUCCESS in 31m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8095408d34bc463699c148b9a8baba2d : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":5},{"id":"e82614b4e4ac02bc0a8885001c02d813a0cd0c8c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-28 11:07:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4812e3f67e99b9d595ddec224e3bd110dc057e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:59:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bce33e2dac2c49d3b82dcd327d12f78c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/16f3bcd5b1c146e7bc093bd3d0625f67 : SUCCESS in 3m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/036c519f5a9d493693fec337c11700eb : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d78dd8185054338b90eca5e4e37dddc : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c67a6c23736a44e8b8d9c142b217b4f9 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb7acf7bac0483482d15f54087472b3 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7418a88e7a234aab8c4e9a9aa18f049e : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf9e4ff55f1c4b7b855066a320944221 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69ad2a75047a4bf39847b36edd8f71fd : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0bd06fbfafd54f05a6e1ad3ead1082fe : SUCCESS in 3m 46s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59764eed5b5242ccb37b910e69593c6e : SUCCESS in 1h 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c82af0d92d8a441aa290ab3f62d5c55e : SUCCESS in 4m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a22a49b8778940a3a6d6a14c6e5f73e9 : FAILURE in 1h 45m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1ece4a696eb4eeba209aad45ce66858 : TIMED_OUT in 2h 49m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1823e7577216496b9735d621e55d2cae : TIMED_OUT in 2h 49m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f00b57ad5e5a40dc9149f638ca631866 : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7636c3e6a714520bd665a8fd3f02089 : SUCCESS in 40m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cae271fe06654f61859a4bbcfc3027d6 : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":5},{"id":"14e7afcb7a285bf299daf772bb9100b206ec05b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-28 15:04:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"676b6656b4f1acf8788000d610aa5e759a69c6b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:55:29.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd175a2d90e94a42ad863d350cfcc8ac\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3898eeb7c09449be90602dc4fcf0374c : SUCCESS in 3m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/be4f0dd047714a539543a48a73a0811a : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e9a3b8fa574cc4a1cbb61fd347fde5 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85ddf22beba64aa89c41b95e68f55483 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8067c87c10a94f7d81f3b9b2f16f0dec : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fbc7945f84449ecae33db42fbbbc9cb : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/851042cb2dc744a5a33b5d340e6a8a16 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fe8101d5e8c4bdc8472e97ef1351890 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/853a0a9554df4988a3af106cb92bc639 : SUCCESS in 4m 12s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fc2a699d2ec440fa64adeb1ca30ab35 : SUCCESS in 1h 17m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db6df77909cf452696c31d438dd3f579 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d9fa93d90fe4513affdd097f7d408b0 : TIMED_OUT in 2h 49m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99617beeeaf64c8b822b125317605559 : FAILURE in 1h 43m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8b488b62499549e0a70ecd303b0e2162 : FAILURE in 1h 30m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/096b7adb81b4426e8e7834b1d4bf3c6b : SUCCESS in 1h 32m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/940dd2600d0a492d8500d166e8d3bd1c : SUCCESS in 40m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d8e98fb67076429387fb63a0323e6477 : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":5},{"id":"413e9672ef85273f1dc95790fde61d9a4230ca71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-29 13:38:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d88a5a4ae37dd5c31cef7214efe00a344d26854e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 15:55:25.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b905dd3865484ca5ac1580dfc8621a6f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/02f7d10d30744cefa5d5e6bae84acfc5 : SUCCESS in 4m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad72f9f484254f7ebd4be937ee898406 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0efff98191f4b3bb11373fd849cd7e6 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/babacf1bfb3e4bc190b1fb8e714d90a7 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a38e62d5284a9988c09c142fc83609 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c45602cfa07c401788774f87745ad980 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3775644e7d4344319ed28ae2fa083c57 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33047978b01045588ece250a366f4e41 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dab7b35370bd4fc0a74c117784ac8095 : SUCCESS in 3m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd750d2445d24dd28779f6ec96660e3a : SUCCESS in 1h 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89bade925de444b8a0ec221d54d64995 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a3aee2522b0c42de9898a18be8d9b6f4 : FAILURE in 1h 22m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d41267a42cf4facaa3d798cdbcae0ed : FAILURE in 1h 58m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f65fd31ed21640cc80477d3bf0ddf529 : SUCCESS in 1h 44m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/29ff91762a48472f9981407b4fd18248 : SUCCESS in 2h 06m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c15342556ec64540a8f53ef3bfef9fca : SUCCESS in 40m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/72a6bd89d46f4e82ab4aff3ac17780f6 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":5},{"id":"e3a86aa1aecc66a86932491efcaa6515d107e1c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 15:31:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"64ccd54287da5b0ec8e65b129f1a2bfd840fc47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 18:23:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4788dc0bce34bb586310dbf1ab8b88b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b5ba90e2ec849cbac4820cc1da30353 : SUCCESS in 4m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0d5a6c471b84824ae032cad8b494087 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb8fa7fa24f410ca556fac130d0eab9 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3016ef839f17440a9302a6bf0982a251 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5573e2287d07432f819a590f18613d98 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d86d8fd2d12435bacf0d3ba5c6d4571 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2210a9431fad47b0a14e4604edcf10ea : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e18ddbf2e8949a7a70b595f8a5fd1a0 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/368f290f8ce64a42a3970716e2336a65 : SUCCESS in 7m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/882170727763433b900f4299bb17e4cb : SUCCESS in 1h 29m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d9f5ab9eb1240b6b534e123139f53a6 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2a03d7ec2aef4d7e95fac68e547eff9f : SUCCESS in 1h 23m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5c1bb7771b14aafa2d7f241c0d5ae20 : SUCCESS in 1h 17m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/353608b1c5d0414f8f308f7b981fc979 : TIMED_OUT in 2h 49m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f509b149baba4f8fb17d01246c3f637b : SUCCESS in 1h 41m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4104ae1a68144f3c8d6d4c5c0acf6133 : SUCCESS in 42m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/846e234a16f74b609891e9fe008eeaf7 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":5},{"id":"43c0842640c7c9b4dc16470ecec09fc560aed3a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 08:58:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf7f82970e6bf328ca85ee1ca5267224fe19f361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 10:57:45.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18fb2968edfd4b55b6d52071f19880b7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/598143a1b49b4c268e4951b057253978 : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6389461e46554230b4b58895560b8bd8 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831bddad77e3474db9b76a0226e35944 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9e22c910a24aebb7521c6c75ed32a0 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b242e4e94544b9d9c62dd27bf918fdf : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/490bf9d9a60443bd8e6da3c18b26e1d0 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ee877ab26dc4acca05d287f626afc84 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4afa7f7bfa744b9b48217a42ba2dac1 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10ec6463344d48b4ad7b30000bb8afd0 : SUCCESS in 4m 12s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9add3c06fd0046fb9bc5805b47e1bf8e : SUCCESS in 1h 34m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a65a365879841588fb00bc418e54fe0 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4bd9bf86b7b449db0e836e87de42d7a : SUCCESS in 1h 50m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22019c2b7f22463bb6184aa51ca13173 : SUCCESS in 1h 56m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9cd08f60175049edb797a3b685855b7b : SUCCESS in 1h 39m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f50b971f455f4ea9a0e7f0e84c6adf64 : SUCCESS in 1h 42m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd4bcfb3a96b4c2f8a63870accf592f3 : SUCCESS in 46m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/694cf73caadc4435bc5b9b68676cc9b8 : FAILURE in 53m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d448f05ab6d37e68b09325b2ca7b6f847107fd13","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 16:43:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"32fda248e901d1e715ef5369d040c750a3b4a3d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 18:39:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70df537280e74a7e981660c348e3c8e1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b6b7fbca72cb40639fb318fe243b3f0a : SUCCESS in 4m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e28f1cb2f92b42589e4a9b8c03cd63b5 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c716dabd4bc43e7b27a74df26f332cb : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc32fbc98e844c6db61c27adfd3d4857 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31095505784c412a91689496652e4d06 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c30e0d4b4a542ce96d6213f0f0f453b : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b43d13e4ca8430e9c6496210f9be02e : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77697d84a2ea46959c44737530124b9e : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be8437b1d1024c47825c52f0dd067924 : SUCCESS in 4m 35s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38afc4acbd9645019188e8f6a2757739 : SUCCESS in 1h 36m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c4c457bdcf149e590107a0e93793298 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e89b72b44ad468ca7e2f5149554680c : SUCCESS in 1h 26m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57cc03b4e0784be38f0a35b3286d953e : SUCCESS in 1h 54m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7a3973cbf69b45debcc5aa91c13ac80e : SUCCESS in 1h 46m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/666913e4816d41b9bfca731b076e8f4f : SUCCESS in 1h 47m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d1844371efc46468a99298a4228595f : SUCCESS in 31m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a019713936047e98b5122314c98eb11 : FAILURE in 55m 12s","accounts_in_message":[],"_revision_number":5},{"id":"07897fc42cf0d5a12fd5bdb965a253b922e5d23f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 14:28:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ffb347617202415e95f128a6a1de239c608e4fc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:46:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/155e5e8b6fd44c37b70454e83c2bfdf6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/09142d6e8f554e41b7bd4a74f5f0434f : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fde7e51a404d4a299637c4601c49aaae : SUCCESS in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e14caad247f4602a8aa1edd7b79f682 : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dfad27bbaa64013b09e548a25538909 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad89f9677224b0faf86bfbdacb0532f : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01f64a6435e245bcadda18254be4349b : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86bc7e82467f46f992c7f5bd7f7eaa73 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f37a9456101f4eba84b7eb59093abcba : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e8071b850774388b8f877491e47742a : SUCCESS in 5m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18c2df2a7b8a46448225e2f9a2efc1d1 : SUCCESS in 1h 43m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1d433a1056b4217b2e83f30f2f4b4a6 : SUCCESS in 3m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73871fcfd2df4a38b5723ddd4c82c6ff : SUCCESS in 1h 24m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e968c63fff6642d09de9f9735238360a : SUCCESS in 2h 16m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3fe76d71b1514f89b5cfb91ecbef0844 : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72502f76bfa1436abecb51602468c32e : SUCCESS in 1h 47m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30042af90e054a1e9945cfb206ede8e9 : SUCCESS in 32m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85004ddcf325441f8c2902a8bf36de88 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f977bdbba961b60ed875ab37e31fecf0ca25da92","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 09:53:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"36d8b229e340a795b5b9d54cbf1a691da08b7bfe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-07 13:57:40.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"881de75e0eec8a8b9e71d24c2cdee88182c0060a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-07 18:38:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ea608029ad5d2f6595725420a04136aa25b15c3b","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-11-08 09:12:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2a441108579225ebb2603a89d8c9d481dca6235b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-11-15 20:54:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"99f6b9b6d01b1c16e02e487201fa87d6083feac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 20:55:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e6ea7b1e2f3e76d4e5a3c58d699d4c54950e2bc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:07:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c69c11f08dc747aa8066b73251665e5f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f6071c6e45ac42a7a670151a0705309e : SUCCESS in 3m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cfc080173c884d9a9064acea2e2a14bf : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b96cce3f7054552b431f3282a8df58b : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad373e7c823475bb1395a7da7d8800f : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246bc87013fe42b091e8c271a01dc8f2 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/356bd2140b7f4ae0b987b46a36fb3647 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c11590ac88424db687921ad80ec04acd : SUCCESS in 1h 32m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/986e5e9a451a48df839f36a105504cbc : SUCCESS in 3m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e8eae9ece19841348b091e1cb3060643 : SUCCESS in 1h 24m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/57f3075970844b0ea07c764e590c736f : SUCCESS in 1h 53m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc56ef1417a4468a871526c97eccf2d6 : SUCCESS in 2h 10m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a25da918d9d244988f22c85bf4f46cec : SUCCESS in 1h 51m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd9f2fee020e429e8a4a1ef458b9128f : SUCCESS in 46m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ec0c3fcab1cfb0139f12b8b0ec58f296b25149ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:07:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2520052982f09a8e1d138b401fac2ea539a39ddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:08:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc7ece634b6d436eb22993815be5e0ea\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9f1c00e9ea004d65b806e42336bc7390 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31edad1d14c7468b96ab83b6b46a336c : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dac9b1951b104c5a9ff46db8ed4ae4a3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"08b8c6d33b34de4be96116ed78835645dca9493a","revisions":{"d9d66b8d57425e22b3e1541e7bf1eeb43a27263e":{"kind":"REWORK","_number":1,"created":"2023-09-20 15:30:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/895940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/895940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/1"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 14:30:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 17:06:12.000000000","tz":0},"subject":"Add the \"abortable\" flag to the ``CallbacksManager`` events","message":"Add the \"abortable\" flag to the ``CallbacksManager`` events\n\nThe ``CallbacksManager`` class considers, by default, that the events\nstarting with \"before_\" and \"precommit_\" can raise an Exception\n(``CallbackFailure``) in case that the callbacks associated to these\nmethods exit with an error.\n\nHowever there are some other events (those started with \"after_\") that\nwon\u0027t generate an exception in case of error. The error will be logged\nbut the process will continue.\n\nThis new functionality adds the possibility of adding any kind of event\nand mark is as \"abortable\". The ``CallbacksManager`` instance will check\nthe errors returned by the callback methods and if any of them is marked\nas \"abortable\", the manager will raise a ``CallbackFailure`` exception,\nfinishing the process.\n\nIn case of being a Neutron worker, for example, the\n``oslo_service.service.Services`` class will restart the process again.\n\nChange-Id: Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865\nRelated-Bug: #2036607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9d66b8d57425e22b3e1541e7bf1eeb43a27263e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9d66b8d57425e22b3e1541e7bf1eeb43a27263e"}]},"branch":"refs/heads/master"},"3ca269098333a64753cc526f8cc5594717c9dbac":{"kind":"REWORK","_number":2,"created":"2023-09-20 16:21:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/895940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/895940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/2"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 14:30:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 17:56:54.000000000","tz":0},"subject":"Add the \"abortable\" flag to the ``CallbacksManager`` events","message":"Add the \"abortable\" flag to the ``CallbacksManager`` events\n\nThe ``CallbacksManager`` class considers, by default, that the events\nstarting with \"before_\" and \"precommit_\" can raise an Exception\n(``CallbackFailure``) in case that the callbacks associated to these\nmethods exit with an error.\n\nHowever there are some other events (those started with \"after_\") that\nwon\u0027t generate an exception in case of error. The error will be logged\nbut the process will continue.\n\nThis new functionality adds the possibility of adding any kind of event\nand mark is as \"abortable\". The ``CallbacksManager`` instance will check\nthe errors returned by the callback methods and if any of them is marked\nas \"abortable\", the manager will raise a ``CallbackFailure`` exception,\nfinishing the process.\n\nIn case of being a Neutron worker, for example, the\n``oslo_service.service.Services`` class will restart the process again.\n\nChange-Id: Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865\nRelated-Bug: #2036607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3ca269098333a64753cc526f8cc5594717c9dbac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3ca269098333a64753cc526f8cc5594717c9dbac"}]},"branch":"refs/heads/master"},"f69d303f8d3ceb5c6eae300363857217d43688d6":{"kind":"REWORK","_number":3,"created":"2023-09-20 16:36:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/895940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/895940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/3"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 14:30:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 18:12:19.000000000","tz":0},"subject":"Add the \"abortable\" flag to the ``CallbacksManager`` events","message":"Add the \"abortable\" flag to the ``CallbacksManager`` events\n\nThe ``CallbacksManager`` class considers, by default, that the events\nstarting with \"before_\" and \"precommit_\" can raise an Exception\n(``CallbackFailure``) in case that the callbacks associated to these\nmethods exit with an error.\n\nHowever there are some other events (those started with \"after_\") that\nwon\u0027t generate an exception in case of error. The error will be logged\nbut the process will continue.\n\nThis new functionality adds the possibility of adding any kind of event\nand mark is as \"abortable\". The ``CallbacksManager`` instance will check\nthe errors returned by the callback methods and if any of them is marked\nas \"abortable\", the manager will raise a ``CallbackFailure`` exception,\nfinishing the process.\n\nIn case of being a Neutron worker, for example, the\n``oslo_service.service.Services`` class will restart the process again.\n\nChange-Id: Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865\nRelated-Bug: #2036607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f69d303f8d3ceb5c6eae300363857217d43688d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f69d303f8d3ceb5c6eae300363857217d43688d6"}]},"branch":"refs/heads/master"},"815710aa694a810849eec3b90a11ed2126038854":{"kind":"REWORK","_number":4,"created":"2023-09-21 08:31:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/895940/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/895940/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/4"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 14:30:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 19:10:38.000000000","tz":0},"subject":"Add the \"cancellable\" flag to the ``CallbacksManager`` events","message":"Add the \"cancellable\" flag to the ``CallbacksManager`` events\n\nThe ``CallbacksManager`` class considers, by default, that the events\nstarting with \"before_\" and \"precommit_\" can raise an Exception\n(``CallbackFailure``) in case that the callbacks associated to these\nmethods exit with an error.\n\nHowever there are some other events (those started with \"after_\") that\nwon\u0027t generate an exception in case of error. The error will be logged\nbut the process will continue.\n\nThis new functionality adds the possibility of adding any kind of event\nand mark is as \"cancellable\". The ``CallbacksManager`` instance will check\nthe errors returned by the callback methods and if any of them is marked\nas \"cancellable\", the manager will raise a ``CallbackFailure`` exception,\nfinishing the process.\n\nIn case of being a Neutron worker, for example, the\n``oslo_service.service.Services`` class will restart the process again.\n\nChange-Id: Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865\nRelated-Bug: #2036607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/815710aa694a810849eec3b90a11ed2126038854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/815710aa694a810849eec3b90a11ed2126038854"}]},"branch":"refs/heads/master"},"08b8c6d33b34de4be96116ed78835645dca9493a":{"kind":"REWORK","_number":5,"created":"2023-09-22 08:26:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/895940/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/895940/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/895940/5"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 14:30:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-12 04:32:26.000000000","tz":0},"subject":"Add the \"cancellable\" flag to the ``CallbacksManager`` events","message":"Add the \"cancellable\" flag to the ``CallbacksManager`` events\n\nThe ``CallbacksManager`` class considers, by default, that the events\nstarting with \"before_\" and \"precommit_\" can raise an Exception\n(``CallbackFailure``) in case that the callbacks associated to these\nmethods exit with an error.\n\nHowever there are some other events (those started with \"after_\") that\nwon\u0027t generate an exception in case of error. The error will be logged\nbut the process will continue.\n\nThis new functionality adds the possibility of adding any kind of event\nand mark is as \"cancellable\". The ``CallbacksManager`` instance will check\nthe errors returned by the callback methods and if any of them is marked\nas \"cancellable\", the manager will raise a ``CallbackFailure`` exception,\nterminating the process.\n\nIn case of being a Neutron worker, for example, the\n``oslo_service.service.Services`` class will restart the process again.\n\nRelated-Bug: #2036607\nChange-Id: Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/08b8c6d33b34de4be96116ed78835645dca9493a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/08b8c6d33b34de4be96116ed78835645dca9493a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
