)]}'
{"id":"openstack%2Fneutron-lib~898828","triplet_id":"openstack%2Fneutron-lib~master~Iba86fe9a1bbf88223b57a45fb89349c6b1858015","project":"openstack/neutron-lib","branch":"master","topic":"bug/1938284","attention_set":{},"removed_from_attention_set":{"31976":{"account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"last_update":"2023-11-13 20:24:26.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-25 07:02:20.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-13 18:18:25.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Iba86fe9a1bbf88223b57a45fb89349c6b1858015","subject":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)","status":"MERGED","created":"2023-10-19 14:50:01.000000000","updated":"2024-05-13 08:49:30.000000000","submitted":"2023-11-13 20:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898828-bug/1938284","meta_rev_id":"6adc9cc52373961bf19299c1cb712951b44d7305","_number":898828,"virtual_id_number":898828,"owner":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 20:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-10-25 07:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-25 08:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-10-30 01:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-13 18:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-11-13 18:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-19 16:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-19 18:44:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-23 19:56:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-24 09:01:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-25 08:17:46.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-10-30 01:25:03.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-05-13 08:49:30.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"df4001e2d8072b6ae0b6f469843b62687b8f36d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-19 14:50:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9462958ce942f93d0a8db272f9cf14e39ee1978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 16:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0e4de5151446cc9535b264f86c7b10\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7629df1f5eb4b9b865c28d11229b49b : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d95fef27aa947dbbf97dd0c352e5560 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb81458d465445fe989c596d2c18e1a6 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508e50b5ec1047b6aa249bc4e0935a98 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b97095f357f4c0bac402e4665b7a640 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffd573b428014f2896a0256c060dd3d1 : SUCCESS in 4m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72351a80b0f34fa0be781a1a2206d168 : SUCCESS in 1h 43m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f0a618de4de44de927b4b696fbddb07 : SUCCESS in 3m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ec417e65ed74c6faa6fe257884c921f : SUCCESS in 1h 35m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/690d47fca4744a29b10dcccb732bb8d4 : SUCCESS in 1h 36m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/697979e16f1b46dc8a810f2d8918609e : SUCCESS in 1h 55m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/297d0de9cc4f424ba38522f6cad8d18e : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1385760464b04055a5fc86b8988502d2 : SUCCESS in 34m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2d8ec5ecce14519a7872923eba4cf2a : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":1},{"id":"760fb5cc598e96a4c6bf56cbdaa52503c249c0c6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-19 18:44:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54947515cfc6cfec5e926d94686f50b6e3356a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-20 08:25:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5d54ca2c2ff378d38742f0495afc928236a58d68","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-20 08:26:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9227c081d14152a84311147b344f9453b9cf8bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 10:57:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88479851050d46f6bfba51957cf533b6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1b176c6b0b8411e92876b59ebb94d34 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67306c1e32684b04b1750b1ea39d240d : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ee18e02ec60417e837c3688588ad9a2 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e4a6d87b584e4797bba20281e3c3ce : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3e4606b31b644a281882f03a60719b2 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2f85de638fe448e8f0b74f52aa61c13 : SUCCESS in 4m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a8a591a91af4dc1b10e46d32ed74aaa : SUCCESS in 1h 41m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88f39e4c086d46a3a9bd413239f706bf : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/17da2552e5bc46ffae80a5904cce49ab : SUCCESS in 1h 32m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7457f89e1ce4c148d57be58ebf45b2d : SUCCESS in 2h 30m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a3f7afbe4113482795f8c6a5ecbb3864 : FAILURE in 2h 00m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d16639c524b44f0a83822c2f8ce099d : SUCCESS in 1h 39m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df2f05f1d0684bec855b23f9dd18c366 : SUCCESS in 32m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/08334f8cfd254d3f9c942f2838d4ec2c : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a507dde6f27d9616fb9ae811f9447a40f7371a94","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-20 12:06:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08c2e9f874bf4895f85cf2ebe2b58fe3cd3d8567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 14:45:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e4fae48cb1344d5acde8e24b80e27ad\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/df500c04c0504edf81e4a2021f03a48c : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738b806caf1a47d38c5393d1034c5a08 : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef89706f42274208a770a4a23626ccc7 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83f306d14ac4258a7098fc89304019e : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f30837141a2c498a83029be3bfb21481 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a45eb6f4c3f84c62b1ac1c87af3f4efc : SUCCESS in 4m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a1e386f12f84dda94904dfcc2d319fe : SUCCESS in 1h 35m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe5a340b316d401380f41116f045b461 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a8bcc2e98f5743a3b0ebfe74fd1e3e1f : SUCCESS in 1h 49m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b65dfb24d2c544d082a8cabe45ff9c4a : SUCCESS in 2h 37m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d1452167fd544fe39de3f16ba1546bd9 : SUCCESS in 1h 58m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e5e131625e2c4e7aa0d78f7d2d53cff2 : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9da92b445dff44ab945c93fba5754513 : SUCCESS in 50m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af57585df7c9471fac6bdfd28e29505c : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":2},{"id":"6b67789027f1ee1d3e438fdf24439880f887825f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-23 19:56:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44742abe28709251abb89fdd7c71f940af9754b5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-24 09:01:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4dab1e3bcf4a958e0d448c927882cbb1b499687b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-24 11:12:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"454734d7273a8c1af1592efdbcc55c177c590f46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-24 11:19:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"594e872da280e56f3dccf46b72f69e0e5161cc77","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-10-24 11:20:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3035b8e5c2cb982bb770c7a53061597a50f5ce60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 13:04:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc5f95b46cae4fa8b882522001fa7a23\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2a8f4e58b5d646dba4008a6aaa490ec4 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296c4661ff7849e4b1d0534260e91440 : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baa34835d9b84151ae076b389ed3f3e7 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae594d05bfd451baf5191110287a1e9 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ba1027c79eb40a48c1c282776528122 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c135f8c5e2d647038c334c19125666ca : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0aa80848f7ed4d788121522d01b6fc2b : SUCCESS in 1h 34m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db223d798254bb4a2b4c0ce89e07c3e : SUCCESS in 4m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72d0d245d7294eb58b8e52fafc50b85b : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3c9198ebba084c228904804ff539994c : SUCCESS in 1h 42m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5471259726f249aface41420627cda49 : SUCCESS in 1h 18m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ea258cad4714773855bf18a95df9232 : SUCCESS in 1h 39m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72f59ba364af452483308906ff839178 : SUCCESS in 31m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6248320ce00e46d6852334c5c27b8316 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":4},{"id":"b622847e93b0d80bdd7886d13ea6f4b358538853","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-25 07:02:20.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6dd6d899461376ce1505ce90ca7fe94a2060b9e5","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-10-25 08:17:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b04a9b7d1076e6258318fa0b3ebcaa15350270a2","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:25:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4282d8c619832bdf3059368ddd35a1aee8bbc437","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-13 18:18:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"131474b0acd4a807c86d3cdc4d055743cc072f54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 18:18:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f44ef890ee39447b7b6bade38a2867b22891bbfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:24:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c04a5315ee546aa86b4114bfdfdef84\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e053ffcd0ec493d92d136203116fade : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf4a6b5f0cd42189fe3855b6fb7ebea : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8546e3b52c4ac1b6447b6bf806c717 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4446c40bf5db4fbdb5de012001d568a8 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b270ec1197a6470bba3b2e965311e5ba : SUCCESS in 4m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6946a2af237a46c5aa3b846f77fb78fb : SUCCESS in 1h 27m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03c694d36545417ba545661fe9036c69 : SUCCESS in 3m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3822dc76b4fb4f9b99116cba7ec95068 : SUCCESS in 1h 43m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2a8434e69067491980a6cfee614e6d46 : SUCCESS in 1h 47m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/302fb5546b7e437890dd987527538cc0 : SUCCESS in 2h 01m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d2188e1e48d2492db094c0dab224cc63 : SUCCESS in 1h 43m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/854bcb7d3736496bbbc4bcb672b05a05 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":4},{"id":"5e5343199e3d1f51f1176808b681292453c8ed5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:24:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"413428dbfb8dd5f2d19c00ff9dfde8186f68d553","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:26:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c23e98e1aae4bd9a4b0ffc12e9c2dd3\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cb9c21ab91064e6c9424de536471415f : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1382e8e04a24315a018cb0632a02202 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97e66ecda3dc445e9a487daa19e5ca6e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b7598dab47d3648bf1bb28eae8c8ed8006097f5a","revisions":{"749f2d54b328105af70b30343b76eaac222ef8ac":{"kind":"REWORK","_number":1,"created":"2023-10-19 14:50:01.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/28/898828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/898828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/1"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-18 11:50:58.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-19 14:49:59.000000000","tz":120},"subject":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)","message":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)\n\nEncryption algorithms: add AES CCM mode and AES GCM mode variants\nfor 128/192/256 bit keys and 8/12/16 octet ICVs.\nAuth algorithms: add aes-xcbc and aes-cmac.\nPFS: add Diffie Hellman groups 15 to 31.\n\nRelated-Bug: #1938284\nChange-Id: Iba86fe9a1bbf88223b57a45fb89349c6b1858015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/749f2d54b328105af70b30343b76eaac222ef8ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/749f2d54b328105af70b30343b76eaac222ef8ac"}]},"branch":"refs/heads/master"},"fff5c8b9523883f370449500195b270db67ca229":{"kind":"REWORK","_number":2,"created":"2023-10-20 08:25:14.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/28/898828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/898828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/2"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-18 11:50:58.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-20 08:24:56.000000000","tz":120},"subject":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)","message":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)\n\nEncryption algorithms: add AES CCM mode and AES GCM mode variants\nfor 128/192/256 bit keys and 8/12/16 octet ICVs.\nAuth algorithms: add aes-xcbc and aes-cmac.\nPFS: add Diffie Hellman groups 15 to 31.\n\nRelated-Bug: #1938284\nChange-Id: Iba86fe9a1bbf88223b57a45fb89349c6b1858015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fff5c8b9523883f370449500195b270db67ca229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fff5c8b9523883f370449500195b270db67ca229"}]},"branch":"refs/heads/master"},"3519de2ccf3f35578adf54d47fcf748d6abe5a90":{"kind":"REWORK","_number":3,"created":"2023-10-24 11:12:00.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/28/898828/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/898828/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/3"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-18 11:50:58.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-24 11:11:47.000000000","tz":120},"subject":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)","message":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)\n\nEncryption algorithms: add AES CCM mode and AES GCM mode variants\nfor 128/192/256 bit keys and 8/12/16 octet ICVs.\nAuth algorithms: add aes-xcbc and aes-cmac.\nPFS: add Diffie Hellman groups 15 to 31.\n\nRelated-Bug: #1938284\nChange-Id: Iba86fe9a1bbf88223b57a45fb89349c6b1858015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3519de2ccf3f35578adf54d47fcf748d6abe5a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3519de2ccf3f35578adf54d47fcf748d6abe5a90"}]},"branch":"refs/heads/master"},"b7598dab47d3648bf1bb28eae8c8ed8006097f5a":{"kind":"REWORK","_number":4,"created":"2023-10-24 11:19:08.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/28/898828/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/898828/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/898828/4"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-18 11:50:58.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-10-24 11:18:59.000000000","tz":120},"subject":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)","message":"vpnaas: add support for more ciphers (auth, encryption, pfs modes)\n\nEncryption algorithms: add AES CCM mode and AES GCM mode variants\nfor 128/192/256 bit keys and 8/12/16 octet ICVs.\nAuth algorithms: add aes-xcbc and aes-cmac.\nPFS: add Diffie Hellman groups 15 to 31.\n\nRelated-Bug: #1938284\nChange-Id: Iba86fe9a1bbf88223b57a45fb89349c6b1858015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b7598dab47d3648bf1bb28eae8c8ed8006097f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b7598dab47d3648bf1bb28eae8c8ed8006097f5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
