)]}'
{"id":"openstack%2Fneutron-lib~900557","triplet_id":"openstack%2Fneutron-lib~master~I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3","project":"openstack/neutron-lib","branch":"master","topic":"bug/2043141","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-14 17:24:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3","subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","status":"MERGED","created":"2023-11-09 20:52:41.000000000","updated":"2023-11-14 17:29:27.000000000","submitted":"2023-11-14 17:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900557-bug/2043141","meta_rev_id":"84a82d18985db3705ec2d4397563c708518b969f","_number":900557,"virtual_id_number":900557,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 17:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-11-14 15:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-11-13 15:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-11-14 15:26:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2023-11-14 13:16:37.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 21:52:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 15:08:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-14 15:26:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"46fc052739ec1c54c5fa65b171e4539f056ec16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 20:52:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa3f665a409fe8962c59f751bf6b6caa659a8cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 21:03:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a34aefbe84affe882874c641a26fa73d7d7cf759","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 21:21:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ecebda75c391a53541d561f3e5a2a8e41588eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 21:22:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9185deeb34fbb2e7cc6af844828e0cac4fc94ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 21:37:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e56ac87530de41ce7f3348b07d6a9f66318e1c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 21:52:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76c5fa36ac61457d83ab68ffa9aa2f8d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9bd43420bf204701805f1a92b4ada842 : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d046c7a6984136bb090df316f9d405 : FAILURE in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca77b3e11117440f85d62b658f75b053 : FAILURE in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d477c21f8a324fe48f921fd39cb2d1a8 : FAILURE in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97c81b3ca7bf452fa9c3abea88bb18bd : FAILURE in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb8439f69bac43eca3d88db1fa0affa8 : FAILURE in 4m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87a75a872b8e473e94ede1f6e4ac2d92 : FAILURE in 12m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/58f5c19f471f4afcb0ab984b50509df9 : FAILURE in 11m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c79505ac062440f78d87cb64ceff46c7 : FAILURE in 8m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4f7d8140eccd495c9a982a509b76de9f : FAILURE in 8m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18b44bf07a7545e4b217b2f4d2b57fe4 : FAILURE in 8m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d77a19df94dc4116861c97c055bceda5 : FAILURE in 8m 19s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3989934e6574428aa90e1359d3be0d6a : FAILURE in 2m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37bbb556311e45f392fb6df9dfa8cb76 : FAILURE in 11m 18s","accounts_in_message":[],"_revision_number":4},{"id":"aa79a64abbad1cb3952c8897aae9113bb4c3753a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 22:24:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d27287cdc4aa9828b4f2b4aadc15c7cefe59d3ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 23:09:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9a9331423dc5cfc52205fefd2ba72ae16229dd71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-09 23:24:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d699ff4709ab18d58df58ecc7fe65076de4b7979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 01:18:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835fcb6d43734a29b05f3c67fb7a9743\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5260b309304a7295f73cd54ed359ac : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b04f8158fb440a88ac813a6241f423d : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9e829d352f4a21a2a62183354c1188 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7f7fdd33fa2478a9e63c6dba8b43fec : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/664db6c6886b4bea84c0c520b834a442 : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4ca8b06a40e47e6bf1ef3d67e42fb2d : SUCCESS in 1h 05m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e5f10eccbd64665a1e8034349894237 : SUCCESS in 1h 50m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/973af1ab3f4c433ca7101695662cfd88 : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2c4f1842b54b42d2a427a67b97f36371 : SUCCESS in 1h 38m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7386c13524324d34982da715b0634773 : SUCCESS in 1h 46m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18b86e2613ef49e0a4a6dc41686c0e70 : SUCCESS in 35m 04s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/129b7fad79624caba30927e09719e2ed : SUCCESS in 5m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48a631ea26934f28b9d931b0f7fd4c47 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":7},{"id":"7adbdc05dd45b58c7dcf7d33f89975b369fce225","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-13 15:08:47.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e68c36ffa0d162f5ad3b3e136a79dfe405a317b7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-13 20:28:35.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ae8b157f98721fc524b310281cf496c5abbff54b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-14 13:16:37.000000000","message":"Patch Set 7: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1ce6861377346d528ab043dabf46c3f149a8b2b6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-11-14 15:26:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cac44bd810097c5ef2ee28c99ac79c9a3d153d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 15:27:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"72d14d4f423d79109cee3e60779044c041db33b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:24:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f33cd5b1c44655aac8cf7e31abae01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec7fb3c6c7e405f82e39ad0b333572e : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7f2587262c4f3eae4e90193b546480 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c4a725cf9094004a929860be448cadc : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/010c70ac7a524ff396fae640a2482b6d : SUCCESS in 4m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d32e7aefe4044aa1ad4365dca82b1584 : SUCCESS in 1h 49m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4717c6030f02485f8e7d43b453664995 : SUCCESS in 1h 18m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b2907bf8d51348a2a4187f089f22dd90 : SUCCESS in 1h 38m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2419b8d022714d219349da8617508de9 : SUCCESS in 1h 27m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c100b2c74c354af7976c91cac62c47e9 : SUCCESS in 1h 33m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb6943b15ca24e72acdf4a5b6bfb90f9 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":7},{"id":"fff8d1b305c493c29f025a50afa430d66552777e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:24:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"84a82d18985db3705ec2d4397563c708518b969f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:29:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80cd207305504532aab24dc4278649df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2948d1ef542947278da011228d40d281 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"825d292c972e1a0478ce965603fed8f9114ee211","revisions":{"9b291cafdba8b24935941066dade39a42414d238":{"kind":"REWORK","_number":1,"created":"2023-11-09 20:52:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/1"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\nCloses-bug: #2043141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9b291cafdba8b24935941066dade39a42414d238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9b291cafdba8b24935941066dade39a42414d238"}]},"branch":"refs/heads/master"},"b19393140eb11fb44e40bd6e0e56410f365ad1d7":{"kind":"REWORK","_number":2,"created":"2023-11-09 21:03:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/2"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 21:02:51.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nAlso added an sqlalchemy-master job so we do not regress.\n\nCloses-bug: #2043141\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b19393140eb11fb44e40bd6e0e56410f365ad1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b19393140eb11fb44e40bd6e0e56410f365ad1d7"}]},"branch":"refs/heads/master"},"9e3c23e205f77da281da391bf5e108579e6f6a0d":{"kind":"REWORK","_number":3,"created":"2023-11-09 21:22:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/3"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 21:21:21.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nAlso added an sqlalchemy-master job so we do not regress\nand changes test-requirements.txt to use a later oslo.utils\nto resolve a sqlalchemy versioning bug.\n\nCloses-bug: #2043141\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3c23e205f77da281da391bf5e108579e6f6a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3c23e205f77da281da391bf5e108579e6f6a0d"}]},"branch":"refs/heads/master"},"4e90321059853e6b44f3922707e2a97bb9587372":{"kind":"REWORK","_number":4,"created":"2023-11-09 21:37:43.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/4"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 21:37:14.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nAlso added an sqlalchemy-master job so we do not regress\nand changes test-requirements.txt to use a later oslo.utils\nto resolve a sqlalchemy versioning bug. Had to move some\nother things forward to resolve some dependencies.\n\nCloses-bug: #2043141\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4e90321059853e6b44f3922707e2a97bb9587372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4e90321059853e6b44f3922707e2a97bb9587372"}]},"branch":"refs/heads/master"},"e9029da18302b78102e07ab66e4add89022a76dc":{"kind":"REWORK","_number":5,"created":"2023-11-09 22:24:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/5"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 22:23:24.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nAlso added an sqlalchemy-master job so we do not regress.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/900517\nCloses-bug: #2043141\n\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9029da18302b78102e07ab66e4add89022a76dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9029da18302b78102e07ab66e4add89022a76dc"}]},"branch":"refs/heads/master"},"b6cba97dd397e4a0743271f70adf1f0cd8c2338c":{"kind":"REWORK","_number":6,"created":"2023-11-09 23:09:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/6"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 23:07:41.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nAlso added an sqlalchemy-master job so we do not regress.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/900517\nCloses-bug: #2043141\n\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b6cba97dd397e4a0743271f70adf1f0cd8c2338c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b6cba97dd397e4a0743271f70adf1f0cd8c2338c"}]},"branch":"refs/heads/master"},"825d292c972e1a0478ce965603fed8f9114ee211":{"kind":"REWORK","_number":7,"created":"2023-11-09 23:24:50.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/900557/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/900557/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/900557/7"}}},"commit":{"parents":[{"commit":"bdebe1de3c9e4032a7a40785846065b676b45b64","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdebe1de3c9e4032a7a40785846065b676b45b64"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 20:49:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-09 23:24:35.000000000","tz":-300},"subject":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0","message":"Change test_sqlalchemytypes.py to support sqlalchemy 2.0\n\nWith the move to sqlalchemy 2.0 some of the API methods\nhave changed, but these tests were never updated.\n\nAlso added an sqlalchemy-master job so we do not regress.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/900517\nCloses-bug: #2043141\n\nChange-Id: I9e1c796dcda725844ea6b7a0d5c8a8c11560b2b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/825d292c972e1a0478ce965603fed8f9114ee211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/825d292c972e1a0478ce965603fed8f9114ee211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
