)]}'
{"id":"openstack%2Fneutron-lib~902537","triplet_id":"openstack%2Fneutron-lib~stable%2Fussuri~I800e0356714d59ba93ab6252c77be0a82f024055","project":"openstack/neutron-lib","branch":"stable/ussuri","topic":"bug/1918145-stable/ussuri","hashtags":[],"change_id":"I800e0356714d59ba93ab6252c77be0a82f024055","subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","status":"ABANDONED","created":"2023-12-04 22:13:15.000000000","updated":"2024-01-17 15:38:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":885084,"cherry_pick_of_patch_set":1,"meta_rev_id":"da6675b0f5edf1928f23fb342226af2aad8e87f6","_number":902537,"virtual_id_number":902537,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-04 23:19:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2023-12-04 22:13:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-04 22:13:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-12-04 23:19:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e7f1ddfa75383387efaa53b1230b1cca7d79ba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-04 22:13:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"88510583745821252f819f0975f1020a1d351b1a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-04 22:13:44.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11d95fef28068b47a17b7d33c9dc220e441264b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 23:19:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d42d706020449a825fe30ac87a40aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f33c4da99e458982b5812088e4a011 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6fba7a7283e4f60b7493f68950efa72 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28e263cb5d7649d3813e77b963bf912a : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4222c798635144dcb9b831c8c9cca675 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f82c17ea1e14703b259d0211eb3a2be : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c135ed5cca64cd7b7f8d7d49a70e0fa : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":1},{"id":"da6675b0f5edf1928f23fb342226af2aad8e87f6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-17 15:38:21.000000000","message":"Abandoned\n\nThis branch of the project transitioned to End of Life. Open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1eb01fe7797847d9d129785638670ab3e039306","revisions":{"d1eb01fe7797847d9d129785638670ab3e039306":{"kind":"REWORK","_number":1,"created":"2023-12-04 22:13:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/37/902537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/37/902537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/902537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/902537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/902537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/37/902537/1"}}},"commit":{"parents":[{"commit":"75ee705ed5f4144483686d7e237ba78afe6d1dff","subject":"Enforce policy for qos_policy_id attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/75ee705ed5f4144483686d7e237ba78afe6d1dff"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-28 15:43:30.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-12-04 22:13:15.000000000","tz":0},"subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","message":"Add a \"GROUP BY\" clause on queries with RBAC entries\n\nAs reported in the Neutron patch [1], this change introduce a\n\"GROUP BY\" clause on the SQL queries with RBAC entries. With [1],\nall resouces with RBAC entries (\u0027network\u0027, \u0027qospolicy\u0027,\n\u0027securitygroup\u0027, \u0027addressscope\u0027, \u0027subnetpool\u0027, \u0027addressgroup\u0027) will\nload the RBAC entries with \"joined\" strategy.\n\nBecause of the low cardinality of the RBAC query when all the RBAC\nregisters are in one single project, this patch groups the resource\nqueries by the resource ID. That will reduce the results returned by\nthe SQL engine to only the singular registers required.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/884877\n\nConflicts:\n    neutron_lib/db/model_query.py\n\nRelated-Bug: #1918145\nChange-Id: I800e0356714d59ba93ab6252c77be0a82f024055\n(cherry picked from commit 829e97024c2b73dd67bfd8a04c65f03be556eec8)\n(cherry picked from commit 4a952047823c3c330dad5475d67f77c7ee8d6419)\n(cherry picked from commit 619c0fe5345af66061730ce763a1cd45c353f37c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d1eb01fe7797847d9d129785638670ab3e039306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d1eb01fe7797847d9d129785638670ab3e039306"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
