)]}'
{"id":"openstack%2Fneutron-lib~903335","triplet_id":"openstack%2Fneutron-lib~stable%2F2023.1~Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865","project":"openstack/neutron-lib","branch":"stable/2023.1","topic":"bug/2036607","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-13 08:20:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865","subject":"Add the \"cancellable\" flag to the ``CallbacksManager`` events","status":"MERGED","created":"2023-12-11 11:59:20.000000000","updated":"2023-12-13 12:32:14.000000000","submitted":"2023-12-13 12:30:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903335-bug/2036607","meta_rev_id":"56396c993f0a7f3f05be672ae3570116e7b9d03c","_number":903335,"virtual_id_number":903335,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-13 12:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-12-12 13:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-12-12 18:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-12-12 18:38:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-11 13:46:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-12 13:00:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-12-12 18:38:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"2297cfbb925d0cda9bbdf19e41baddf6d7337d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-11 11:59:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54015469281d329903e658e0d623cdf9f848dee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 13:46:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc7952f9dca415ba9c850a56a560493\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9faac2cff33f44269d76da9548171e8a : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac3f2e70b864f87877f087b090435a5 : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a95f9473cf164e02bfe7dd70058fa8a0 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6beb56a338e84a858e297ac4c9226e93 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f6e433d22524200959f3a2e904f96f5 : SUCCESS in 3m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cd0d9bfbb4c4ed6814cdb89069362fd : SUCCESS in 1h 41m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5a1978054ca446cb8d7e515cad5c610 : SUCCESS in 3m 16s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/f5039b64a43f460fbbf656e07bea947f : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/5e4d058a3d244813acdd909367b43b9c : SUCCESS in 1h 17m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/141c705b4f374cc6a4341221ddc386a5 : SUCCESS in 1h 21m 40s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/2cbbd0deb6a84935a39ae0cc361a1929 : SUCCESS in 1h 22m 49s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/98884c7a2a774f5b89ae15b0985c7aa9 : SUCCESS in 34m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9ddde3719f14376bbbf4457b41134e4 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8d12170520f77b91b633d2e33c724f942b43ceb4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-12-12 13:00:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a1599395dfcd26640de0965b709f34630c665215","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-12 18:38:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f425093a2142ce92056585a6ec5305564709bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 18:38:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d77037f6efb882b4ac5cfbae81f74dc034552d0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 20:50:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f516a1c7952a4c8ca186c0295417dfc1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c93a26edf534f50a40924bbce6bfe7f : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ea4e7646c54f15a1e359d19997c35b : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea24c7cb94e24c3fbccf42785773e17b : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51849511cdb34e4cb67a20e4138ad3bc : SUCCESS in 4m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e426bdb882424a009ea2b328ea7c2d18 : SUCCESS in 1h 19m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b00b8937a554016aa080cb4ae7ed5b4 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c966fb2a589f48bfaf7eeee45669876b : SUCCESS in 2h 09m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbb2e8b2f12e44d088fa99a0c150ae76 : FAILURE in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"64343d957424947a4baed36d6697d9c0a5e7c9b5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-12 21:00:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8acdcabb7f7b3a1ea424d35429d28d8de0f1fc64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 23:08:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da83d1e373044504bf3200436ad08089\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/da3db545c0e0461cbe0796a063003fea : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5bbd66a47f46dd8c9827298f761f1b : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc3288461d84f8bb71e2dc91b8bc151 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f7e961d4cf4d42918a0427b5064fe2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16708e7e71564e5b85ea80a9b57eb37f : SUCCESS in 6m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f6759931f764947a3d76638e496fa86 : SUCCESS in 56m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc11c6e1fee43f59528f639be96cbec : SUCCESS in 3m 20s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/9984978415f948a0897e3c5514fccff0 : SUCCESS in 1h 55m 50s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/4a39d4bb7c114431870cb0f9d3cd1b03 : SUCCESS in 1h 38m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/f534db32ef854ae0bae9fc5c7efd3b98 : SUCCESS in 2h 05m 04s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/3ff5a3f5e351494b874b32d97a387e31 : SUCCESS in 1h 29m 25s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/7a56b0a17b074283b9b7740f0517d0c1 : SUCCESS in 40m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84857cadd1a344c0893222bdfd2a6d86 : FAILURE in 36m 39s","accounts_in_message":[],"_revision_number":1},{"id":"50f00acca3a252addc061d857fc34dd0d131be64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-13 08:20:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac5fc40833319fa194c80d59038aed3ac784fa75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 10:26:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/551a7685b69646bb846207650d805e6d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/721edde03c534dc5962ae3839851a5a8 : SUCCESS in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae5df5ce0584301b1a1859b378bdcc3 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd0e90fab69047e5b2b6e8ed82f65007 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56944dcc0f64ce499f79a50c6ea18f9 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac6292be571547278cbfe57bd28447d0 : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d8c98da78814b1f91eb6db5552177d7 : SUCCESS in 1h 21m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6f68a12d1574092bd00aac1db6be9e8 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/00680a3c6c0e46b581d9e00f20dbe690 : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/8de5f16b55504e50b3660086a7f2fd34 : SUCCESS in 1h 10m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/26a26b71ece6491297bfdc78e93b4e37 : SUCCESS in 1h 54m 56s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/93b679312e0a4b8bbae2801db6f057bf : SUCCESS in 2h 03m 12s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/ff6fd38505a54d938c38707404e1bc1a : SUCCESS in 38m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/205461443a3645518151ab39cbb2cdc9 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"53d53d45f94bb165c32c0c235af20dd9ff788a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 10:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"689c33b21956360d4a34ab2a33bb04f0881d9114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 12:30:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2830dddd1b842ada6b2d2920ca44e82\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81b3b5245e854f40a0b489c05f88c0a3 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a545f767edc14c71837bcd1d27e1788d : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c3a807d3cf4f0186305065b783db76 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c981a7f1cd547b49adc2afd06f3bfe1 : SUCCESS in 4m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6bed8a577a14c62ab5aeb75786f19ce : SUCCESS in 1h 16m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6dab7e625fd40ca94f0836383a381d4 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/a02bd76397e64d2c96dd0ef08bab2a21 : SUCCESS in 1h 49m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f38f8b19b4e4d589a686b51369e01a9 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fd4fa3da6426bfd463840b3f687cfe980e28df47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 12:30:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56396c993f0a7f3f05be672ae3570116e7b9d03c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 12:32:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f31588c77d04927a145fe82c3930ac4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1b79972d0d4d3b9b8fc83358ef75b2 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de5a654e979f4913832dbfbcc4df3e0a : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e0f5983ccfa9b0c04e17a01876eed10187ec1ee","revisions":{"1e0f5983ccfa9b0c04e17a01876eed10187ec1ee":{"kind":"REWORK","_number":1,"created":"2023-12-11 11:59:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/903335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/903335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/903335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/903335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/903335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/903335/1"}}},"commit":{"parents":[{"commit":"86ea887de64bba3d48a245884aea961b04afad0a","subject":"Add a \"GROUP BY\" clause on queries with RBAC entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/86ea887de64bba3d48a245884aea961b04afad0a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 14:30:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-18 05:48:49.000000000","tz":0},"subject":"Add the \"cancellable\" flag to the ``CallbacksManager`` events","message":"Add the \"cancellable\" flag to the ``CallbacksManager`` events\n\nThe ``CallbacksManager`` class considers, by default, that the events\nstarting with \"before_\" and \"precommit_\" can raise an Exception\n(``CallbackFailure``) in case that the callbacks associated to these\nmethods exit with an error.\n\nHowever there are some other events (those started with \"after_\") that\nwon\u0027t generate an exception in case of error. The error will be logged\nbut the process will continue.\n\nThis new functionality adds the possibility of adding any kind of event\nand mark is as \"cancellable\". The ``CallbacksManager`` instance will check\nthe errors returned by the callback methods and if any of them is marked\nas \"cancellable\", the manager will raise a ``CallbackFailure`` exception,\nterminating the process.\n\nIn case of being a Neutron worker, for example, the\n``oslo_service.service.Services`` class will restart the process again.\n\nRelated-Bug: #2036607\nChange-Id: Ie1e7be6d70cca957c1b1b6c15b402e8bc6523865\n(cherry picked from commit 08b8c6d33b34de4be96116ed78835645dca9493a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1e0f5983ccfa9b0c04e17a01876eed10187ec1ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1e0f5983ccfa9b0c04e17a01876eed10187ec1ee"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
