)]}'
{"id":"openstack%2Fneutron-lib~903531","triplet_id":"openstack%2Fneutron-lib~master~Id84f3f5c83a1452875f2d7adf4611db319c240ce","project":"openstack/neutron-lib","branch":"master","topic":"bug/2052786","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-02-22 18:21:03.000000000","reason":"Change was submitted"},"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-02-22 18:21:03.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-22 18:21:03.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-02-22 15:57:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-02-13 15:01:23.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id84f3f5c83a1452875f2d7adf4611db319c240ce","subject":"Add \"socket\" NUMA affinity policy","status":"MERGED","created":"2023-12-12 18:57:06.000000000","updated":"2024-02-22 18:23:11.000000000","submitted":"2024-02-22 18:21:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903531-bug/2052786","meta_rev_id":"9e6630ce5738fba8286496571609136c6fe12773","_number":903531,"virtual_id_number":903531,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 18:21:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-22 15:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-02-13 15:01:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-22 15:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-12 20:46:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-12-12 22:51:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-13 08:14:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-12 08:26:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-02-12 11:03:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-02-12 17:56:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-02-14 07:57:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2f0cf02758ff42eb91f77c55dae458a7a6c386d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2023-12-12 18:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cf41d516ababc22989f45ef2953dc4868f3e545","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2023-12-12 18:58:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a8cd931bce4941a34ea931d57e48a9694ba543e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-12 20:46:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80e11c63d90deebda182ec5366b6fb42bb4f08b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2023-12-12 20:50:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a43fe437dcac7fff8d0760bcf54c617f2ba85ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 22:51:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e1f4fc3dc844fa919e12ba01911eae\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9a4d05ed5b1d4f04b9b3ab294ecbf1fa : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf5777394e546529762604038cbdf93 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/076f32c875054dccb00ebffb63c31f4c : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6714f8e0993a4a739643824a677a06cd : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e190940800934b889b18c5a262957cde : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da837359983b4a719dc816389c586f82 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41abdb82a757410290f1bf32f99fa805 : SUCCESS in 1h 34m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/501ae07039074a23891e179f9a40d291 : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e6521017784409b9f04c223ef7e812d : SUCCESS in 1h 58m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1ff299c330ae444b89ef54b7fd6e3303 : SUCCESS in 1h 50m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c686ec36fea54c1483e57eca9efc93a6 : SUCCESS in 1h 42m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b806b5c6ff8141529d3cf1eddd8c0194 : SUCCESS in 46m 22s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/77467b86a3fb4f1b8bcfb29779f8fed0 : SUCCESS in 8m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9b1e7fbc3864663b4d2150c58d81d59 : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b83ee209834b66be85db46a9bb404d560c59399e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-12 23:06:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a9bceeb44299c81aa7110de3c04f7736f48bd6f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-13 08:14:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3cb5d5a13ce9566796786aa791509fb6f61ab1f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-13 08:18:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d08dbfcc4378615a43d916ad0e8648bfeb8e70d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-09 10:34:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a6b9629efd790b85cec81db692710db5d3f0644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 12:36:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1638943cad4b4bbcabfe216d729084d4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1667cad9fcf54b18b89656a90af8855c : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f4fc53653d49e7a25f3efc6b910614 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7236bf620daa4aec97ad8ae1d6aa7f43 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11fee2682684956af019b8e5288fda5 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08fd0b2637df4a48a5ded567a79f6f7b : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f215f5201ab147a6ae466ebec47e8e2c : SUCCESS in 4m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a19432d2807040df8e5834539349278a : SUCCESS in 1h 35m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51f8f77a9c44e6a9a5454766e261d06 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f54cc5eb0f1b4dc880a165ddef0170d8 : SUCCESS in 1h 36m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f60169eb5c6f47a58ff2442dc777e74c : SUCCESS in 1h 58m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8fad9c9fa90d4fac9e7cee9fd5490eca : SUCCESS in 1h 13m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7debf54c16614b70bcfaf1cb62aba2ea : SUCCESS in 1h 48m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4dcb7cab3724c2ba3f0d168e12d5a66 : SUCCESS in 31m 54s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/34b50f0947dc4df1bdf47d189da7cf5f : SUCCESS in 6m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e57a73f0b21748ac84dd7b83c04ac141 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":4},{"id":"7172b592254cc9c568267aa73edc8285a9ab6334","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-12 08:26:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c7b5b86e6ac4ced075bb57660c3f04245ee532c8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-12 11:03:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"88337d51be0593e6f18238f76bf4a9fdeb2dba7e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-12 17:56:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"347e931489c7b8530c032715dd51ccb9ae3dd7e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-13 06:56:20.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d97dc324071c90961e2c588cf598b04aac76115b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:38:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58be54a9693445bcac6304bf5f07e1bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9d320673a67d4405a0b5e79fab4f89b6 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cdde3ec2abf4846bc09b06831d93f6b : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd0d76f0559b4d4399b3a735cc60676b : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37660df6cafd4137ab0789f9a408d8a2 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2f2e31aeca943e391e09e0fa395575a : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ac8fdbad72c44d7bac5250d36a5e4e4 : SUCCESS in 3m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/000d2935c599428c959ef05aee0478fe : SUCCESS in 1h 17m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24146b5688c14fa190d658813df7e241 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/822d642422bb4ce1880effd8e7688a33 : SUCCESS in 1h 16m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d938f749eca47e1b31f3b868324e0b0 : SUCCESS in 1h 36m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4d209dd83a97490c91a8977b00acc781 : SUCCESS in 1h 30m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/187f8ace54b644848e0f0ab9e31aec86 : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22b980b89fd44bea8b4cb0e6c2a6395f : SUCCESS in 33m 16s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2f10fd14f5504e4091adfefd4f72d750 : SUCCESS in 6m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9df5b58ecb0f40c9a5b3677793fd77c4 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f4246df85f22ede696c503105a7a8106e0d3a9dd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-13 14:06:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cedf3b146100deb67c9669d674877bfdcbcc0329","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-13 15:01:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a5c996fa91a732bbe2605e46bea44632efc58a88","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-22 15:57:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"614ce0166b8c908df7f60d07c751c126d122c9a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:57:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"85ce756e3acf999c19ab8229d62c5603f585d740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:21:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4674838d23bd4ca9a1f40dbb0f727633\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ab6afbb187454b26834a7ae618b2faa2 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adedb20b7c454c4fa683cceb07398d8a : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5962930e02c74fa08214cf7412645334 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3afdecbaf74c4d10a1027b739e1f675d : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d0e7f22bd1f41f1b168e305da320b34 : SUCCESS in 4m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee897b8398194ca797cd970a2fdc77cd : SUCCESS in 1h 28m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f73cd07cb9434d24a8710b99f48cb819 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a76dc2c37a2c4d78b07215cc72a2ec40 : SUCCESS in 1h 50m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb73596941a64ab68fdd32ed20971dec : SUCCESS in 2h 22m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9669d93b3d654dabb504d1bc1a38235c : SUCCESS in 2h 02m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/31d81a9aeba840f68a8914cf5a0bd568 : SUCCESS in 1h 48m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d087b794806242668f69f05e0ba21c7e : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":5},{"id":"c4129b31dc34b30ea059a4cde0b737fdcb164be0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:21:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9e6630ce5738fba8286496571609136c6fe12773","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:23:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d2d87defee9413fba646cda491ce9de\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e1034af96c1d4492b78be1a39c8f1074 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edcbfc9fc3c4e3c9ba6bea530ac9b7c : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14c984586a7e449cabdc6667a42817f5 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7eeb1b87acfdb2131335680f1c0ff55e55c2f2cd","revisions":{"235b55abe7e473291f9e9a6a9287bb9fa9480d9b":{"kind":"REWORK","_number":1,"created":"2023-12-12 18:57:06.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/31/903531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/903531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/1"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:55:01.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:55:01.000000000","tz":-300},"subject":"Add `socket` NUMA affinity policy","message":"Add `socket` NUMA affinity policy\n\nThe `socket` NUMA affinity policy has been supported in Nova since\nId94fdc6789b2bae735b0d1e9a02dce60a8f3fb40. We even have functional\ntests for it with our Neutron fixture, which puts us in the weird\nsituation of the Neutron fixture supporting more things than Neutron\nitself. This patch just adds `socket` to the list of allowed policies,\nenabling the existing Nova support to actually be used.\n\nChange-Id: Id84f3f5c83a1452875f2d7adf4611db319c240ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/235b55abe7e473291f9e9a6a9287bb9fa9480d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/235b55abe7e473291f9e9a6a9287bb9fa9480d9b"}]},"branch":"refs/heads/master"},"e26b399c66bf44935c2367845b9e01483f52e560":{"kind":"REWORK","_number":2,"created":"2023-12-12 18:58:21.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/31/903531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/903531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/2"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:55:01.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:58:17.000000000","tz":-300},"subject":"Add `socket` NUMA affinity policy","message":"Add `socket` NUMA affinity policy\n\nThe `socket` NUMA affinity policy has been supported in Nova since\nId94fdc6789b2bae735b0d1e9a02dce60a8f3fb40. We even have functional\ntests for it with our Neutron fixture, which puts us in the weird\nsituation of the Neutron fixture supporting more things than Neutron\nitself. This patch just adds `socket` to the list of allowed policies,\nenabling the existing Nova support to actually be used.\n\nChange-Id: Id84f3f5c83a1452875f2d7adf4611db319c240ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e26b399c66bf44935c2367845b9e01483f52e560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e26b399c66bf44935c2367845b9e01483f52e560"}]},"branch":"refs/heads/master"},"f6d23cb559c0656aa9cc9431c3aa4a3a294badfb":{"kind":"REWORK","_number":3,"created":"2023-12-12 20:50:46.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/31/903531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/903531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/3"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:55:01.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 20:50:39.000000000","tz":-300},"subject":"Add `socket` NUMA affinity policy","message":"Add `socket` NUMA affinity policy\n\nThe `socket` NUMA affinity policy has been supported in Nova since\nId94fdc6789b2bae735b0d1e9a02dce60a8f3fb40. We even have functional\ntests for it with our Neutron fixture, which puts us in the weird\nsituation of the Neutron fixture supporting more things than Neutron\nitself. This patch just adds `socket` to the list of allowed policies,\nenabling the existing Nova support to actually be used.\n\nChange-Id: Id84f3f5c83a1452875f2d7adf4611db319c240ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f6d23cb559c0656aa9cc9431c3aa4a3a294badfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f6d23cb559c0656aa9cc9431c3aa4a3a294badfb"}]},"branch":"refs/heads/master"},"704ec048d356642d2f2921767cfec577e4fb62c9":{"kind":"REWORK","_number":4,"created":"2024-02-09 10:34:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/903531/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/903531/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/4"}}},"commit":{"parents":[{"commit":"35455d101bb09ca71620216f9b573baa5666a961","subject":"Merge \"Replace CRLF by LF\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/35455d101bb09ca71620216f9b573baa5666a961"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:55:01.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 19:29:04.000000000","tz":0},"subject":"Add \"socket\" NUMA affinity policy","message":"Add \"socket\" NUMA affinity policy\n\nThe \"socket\" NUMA affinity policy has been supported in Nova since [1].\n\n[1]https://review.opendev.org/c/openstack/nova/+/773792\n\nRelated-Bug: #2052786\nChange-Id: Id84f3f5c83a1452875f2d7adf4611db319c240ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/704ec048d356642d2f2921767cfec577e4fb62c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/704ec048d356642d2f2921767cfec577e4fb62c9"}]},"branch":"refs/heads/master"},"7eeb1b87acfdb2131335680f1c0ff55e55c2f2cd":{"kind":"REWORK","_number":5,"created":"2024-02-13 06:56:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/903531/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/903531/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/903531/5"}}},"commit":{"parents":[{"commit":"35455d101bb09ca71620216f9b573baa5666a961","subject":"Merge \"Replace CRLF by LF\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/35455d101bb09ca71620216f9b573baa5666a961"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2023-12-12 18:55:01.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-02-13 06:56:20.000000000","tz":0},"subject":"Add \"socket\" NUMA affinity policy","message":"Add \"socket\" NUMA affinity policy\n\nThe \"socket\" NUMA affinity policy has been supported in Nova since [1].\n\n[1]https://review.opendev.org/c/openstack/nova/+/773792\n\nRelated-Bug: #2052786\nChange-Id: Id84f3f5c83a1452875f2d7adf4611db319c240ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7eeb1b87acfdb2131335680f1c0ff55e55c2f2cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7eeb1b87acfdb2131335680f1c0ff55e55c2f2cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
