)]}'
{"id":"openstack%2Fneutron-lib~903971","triplet_id":"openstack%2Fneutron-lib~master~Icda2da71135065a1192954f33943b9ee225c5cf4","project":"openstack/neutron-lib","branch":"master","topic":"bug/1938284","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-01-15 07:12:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"31976":{"account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"last_update":"2025-01-16 03:11:59.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-14 21:31:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-01-14 15:01:35.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-01-16 03:11:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icda2da71135065a1192954f33943b9ee225c5cf4","subject":"vpnaas: add support for AES CTR","status":"MERGED","created":"2023-12-19 12:10:17.000000000","updated":"2025-01-16 03:13:11.000000000","submitted":"2025-01-16 03:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903971-bug/1938284","meta_rev_id":"104926eba22194a8c22fdc53fa1bf7b1ff520186","_number":903971,"virtual_id_number":903971,"owner":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 03:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-01-15 07:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-01-14 21:31:54.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-01-15 07:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-19 12:28:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-12-19 14:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-08 17:41:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-09 07:13:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-01-09 13:04:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-01-10 13:11:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-13 08:49:30.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-10-24 08:03:33.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2025-01-14 15:22:44.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"}],"messages":[{"id":"0bbe8b554042f43c984b427a1dacba6583e3b547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-12-19 12:10:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"985d069a2ce538a0176487485c9dc8a436453c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 14:20:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5d89a52968a4a919f27e1719fa984a8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9e3a3339e7454df7a33fc046e7367ca1 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c33184954a54111949aed2c0afb212a : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45f048c10cbd4abf91ce2cbf3622b9f1 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f523d6fce6b24d5e8e82a02d2d25a413 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f9ce6b7e374424dabdcd85331bca562 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17b59c7365a54dc387bb5253d5e3309a : SUCCESS in 4m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9ff5f6f99444f09992b49e0d0893ff7 : FAILURE in 1h 12m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6a1a4528d294691b75f2101db73e850 : SUCCESS in 1h 20m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31178c21e7154ea38f11d8bc829aa593 : SUCCESS in 1h 25m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6c701865d4b24b74886728374b93f86d : SUCCESS in 2h 07m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d4b2ba337d54f73901b48b4e9a783d0 : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1735da3655da49e0ac8eba6a526ad839 : SUCCESS in 46m 43s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6d7593cd72bf441c8ea0317243856543 : SUCCESS in 6m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ac5b39f360b4722a2c3ccfceae31f4f : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":1},{"id":"599199b7eb1b4187335dd2833561a9397a1c9c6a","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-12-19 14:22:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93ddfb73128f2da9f117f831908469cc0e533d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 16:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c43561ca81405596ed8e9e66adf12f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/efd7a768f5cf4807bbd06ac7242de062 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea09fa6e0f4443d29cbb42fb612ae9aa : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a55778a5a5468487e182db2b05e474 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edfc901c293e43159391ff140a902508 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d67e4fb88fdf4cada86e00cf571fc887 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/081683da3c8446adb87215d9110673f2 : SUCCESS in 4m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdef7df8ad0a4462a3d0547bd3c56118 : SUCCESS in 1h 26m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b6f0d4dd32141fe84ab29b87fa85df5 : SUCCESS in 1h 55m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41fdce90be3a4410aa46ecfb6f4580f1 : SUCCESS in 2h 02m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4cf171789f0e44efbc3672d3001e68f9 : SUCCESS in 1h 41m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7761b933b37149739fa0284d21835aac : SUCCESS in 1h 39m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57c878ce85f54404a495b57ed8cfe347 : SUCCESS in 32m 42s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dd74fe39212344829a00a1cc179c10a0 : SUCCESS in 6m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2be94d31369a466e9858d22d6370ea41 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c4f5b6a5f6442c05279f351bcd125e8bb643d670","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-08 17:41:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"98c27a16b0ea439acd0eced705fe6aba93b3ead7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-01-08 17:56:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"80618c7e4433dd2627fcc9f676e804ed028c92bd","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-01-08 17:57:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a4e49b01fde8cef6b57291cdc2c470689548443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 19:46:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7f881d8613f437fa77a1177553482bc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5645aa23bf614c8c9a1bd5410113fb2d : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0148ffefef47b99969057c96e6f997 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/081bf7e160f04e998824ed0583a6adf8 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5347f3d5e94c1881237969d1611854 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/084b0a78086945679fa49542a6070c34 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a8d82e556c64db28ced1ef5348653a9 : SUCCESS in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f236bebe62e4bd8a162db7cdc738df0 : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4453370cc69d428f96bbd2b51616981b : SUCCESS in 1h 17m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3f68644047054b3a861bd4b97d07a138 : POST_FAILURE in 1h 40m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a949829d57f4425fadb581f04f8bda8b : SUCCESS in 1h 43m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f1f8c0e3863e46d3a71f36b12bd07000 : SUCCESS in 1h 47m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8ca294e1050e4e12bdf2ccf3db0d20ee : SUCCESS in 34m 31s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/055029f964164654b1b18e20e7d8d2ee : SUCCESS in 6m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a74f89bab0e42ff80a3d0bceb4a42b4 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4f9a9d07eb5c7737a9a4ed54c799f7d9c63367e2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-09 07:13:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f65c43d37cfa2da3ea4a54f7a7906f496b18f9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-01-09 09:13:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d3c9108ef8e69d95300fa6896c1463b9ac8a16aa","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-01-09 09:16:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8d1233b46c695c408db6b0217b27e9599515325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 11:30:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee151d69aa164eca9eb6383e715ae60f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3a0f4712c23e4c268caaf317ba3a53e3 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f101d16f7d4fd3b385db2f8a45f93e : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5188004e36024cc1970f00ab3270a089 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14eb81e4e7a74bc7a12653e7ee3e33a1 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01a1bbe6c144453fbe4c1148be353b9d : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d62701e1fdb14549bc8be7b8dfb09fc9 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dbf34e26d0d47b4bce2db86bbacf1f0 : SUCCESS in 1h 38m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c7eb279f445a49fcaca201831743a739 : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed1e2d729ee54edd847c66be57c1070f : SUCCESS in 2h 14m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3b1a131f3d2e4a94a7a8b2d0dc97d12e : SUCCESS in 1h 19m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/677b93d115a448f290e00f0d5fb25ab3 : SUCCESS in 2h 04m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35e3d6d0dd074d82b1d384dc2f41654d : SUCCESS in 35m 18s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/770bfa9a272941f99581fca15fad1eab : SUCCESS in 6m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2b6a65ee76f4f309ae3e1d2d50f8007 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3976ad641588355e389c2a2ca0dbcef0e5f7abe6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-09 13:04:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65929b805a1173aa63749775c719896207c659ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-01-09 15:58:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b6e299cf2a6953057652373c8a57a0cd1a6e9fa1","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-01-09 16:01:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e79d61c8b47cbd694e742557be12bb49493ac60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 18:23:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/241ef01c9f334cffb891e5c269728a9b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a4cf2ccd6f36402ba64881d18c1cc55d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4f836e96e0489197070e349c33df33 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bba25dcf7be84a80a810006f600f4aaa : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e153d8b00e2c4d4c853f78e6d7b57cb9 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf100b571fc14c938eb06dfcc2fafa66 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e36b732fca64df796b76904a94b00c1 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58668fe09c9048d397d35da2bf34ddfd : SUCCESS in 1h 51m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6cee38b2e9d4e63a64d9be340d84a9a : SUCCESS in 4m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/368357bfc79146fc8e62949c8239646b : SUCCESS in 2h 10m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc7c3b95977541eca2405757d2491a10 : SUCCESS in 2h 04m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fd92dc1eb59941e3817fcb96a821a8dc : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c50a1551d5b4f7eae0c9bc02759e90e : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c93009f3e9044ce0ac6248d393ce9aaf : SUCCESS in 33m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2d346d05d8fc406da512be73132b61a1 : SUCCESS in 8m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cac7c2a570b04e069e7b4d0fa912af6d : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":4},{"id":"f28e45d21206b07411f4d9de607d75a91821d2a9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-10 13:11:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b7e9c60e229c9210d126293d6b443a8fc5809f65","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-13 19:37:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f91e3f3f76058490ee0c07c492a9a53f8ffc124c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-13 19:37:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"48fca388fb96a0a3b5ee84e634edd4509a58e6c4","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-14 10:03:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a74cfc0ee50f5397f6ca1f578b7c2baa5493bb2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-16 07:27:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52637657f8c5f52570e49c5f275f7b76c362d7ac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-20 15:07:16.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"ee36ffab8f7ba1e6a4e9c68ac19690d07506508a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-20 22:18:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ad1e0853ecd71d54b42e4b00219639e3309ba0fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-21 07:10:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ff7249a479dce49211a53bfe16acbdb5097056b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-26 15:27:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d2a2ec3bf701001247e3a2f50b509b3da8fac24c","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-26 15:30:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cc27ed9ad02d2c3f58192af389c87f54aebb6e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 17:09:59.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b21847526cf440ea6a12f8825c8ca91\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd7cbe449e9a4ddeb8e5683e50d6ce6a : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95fe792054b74e41aadd077f2f029a50 : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0bcbeef662d446a8afcd39d7866fef8 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74e91f2332ff402480a49bcc9e5ec10f : FAILURE in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdfc24e7cd754bcd9ff9b5271cd2068d : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad68c740922c4d3f9b5a654b1b9d6751 : SUCCESS in 5m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae64a827fcaf40088c9f8118eab05f36 : SUCCESS in 1h 22m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f74660ff29f541a08d8b309468ec12a7 : SUCCESS in 4m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/712b854e44624e61a2ab5b78f0a5844a : SUCCESS in 1h 38m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1638576a34e74245a25b574a24475649 : SUCCESS in 1h 41m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/11420404741c498caa6e21d8f4bb2c97 : SUCCESS in 1h 14m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52e1394a712e46988842ca02e8a490be : SUCCESS in 1h 22m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8e1b4f280dc402caff97c60782eeeb6 : SUCCESS in 33m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5df3fcb9ee5e4ccc8407f38d2c77ca3d : SUCCESS in 6m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f611628d7f0462895d1f3415ff8ba82 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":5},{"id":"6f832a3a0df3839016bccc913b0104b43161a399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-26 17:19:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c4d9d3a999f4a30b6b6367ebca0e48072fe3dc19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 19:20:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbada01567834f1bb6f637e75c377918\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b3237cba19944eb98a6d58bfb8572346 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d539fc5ae84e40cfbc8d2d125165a597 : SUCCESS in 7m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d678e32e912547449daf3660afcd9fc5 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ed5812793545018433a070803404b5 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc44ed8dfde74432a150059481c126d4 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/216771e28daf4c2ab2da785e0e3c5e55 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99da16f30d7c49208d735805f8549ced : SUCCESS in 1h 27m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d593acae50b34ac69ba959213ada45a3 : SUCCESS in 3m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c85c31448d4f4dbca92cc0ea20835e98 : SUCCESS in 1h 47m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab46b6fb75fd4f6b849804039f609910 : SUCCESS in 1h 59m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/499c430a1b0945fda566f4284163dc08 : SUCCESS in 1h 38m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/73f4f6b9af9840198b6f52dcd4184d83 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45843237a4454f9f86f343ca59144450 : SUCCESS in 49m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/850334cfb4914a799ae8a5d13a2764ff : SUCCESS in 6m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cee1f0c9e1b54bb5ab0ddc68fabb4b17 : FAILURE in 51m 25s","accounts_in_message":[],"_revision_number":6},{"id":"5278cf22766f4339ec6dd3dc28740696d880861e","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-27 09:04:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"385744bb5266b991cd32f06f96b04d2f3e62e1b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:18:17.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50ac4725220d469f9ef7acd070293b75\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/73f79367104d40eaa3068d26f618e399 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f0c2cdde3e4e629ce9f7da77536504 : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccb0f73b14884f01a8961a41af3d2847 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6691212daa9471bb3c509c1781dd536 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f852c31b4c84b36bae9594027371a29 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aff36ece3fd54302a4f0dbf5fc18dfa8 : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f95e50914b64c958785cb4084319d1b : SUCCESS in 1h 29m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6bd2266eeb44599af5f4dd194b54b9c : SUCCESS in 3m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0272ff0413b94b8aa880cea4eac7bf1e : SUCCESS in 1h 29m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd96aac38c3f4a5bb6c7cae185fbe663 : SUCCESS in 1h 45m 42s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7269ccd211e04a07bd87c22210ed6d50 : SUCCESS in 1h 46m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0eb8a5b775ed448ea45964c444385e7b : SUCCESS in 2h 07m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/941e5744c60b48cd8d38786bd6981e4b : SUCCESS in 35m 27s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/327932c11b614a8094dcf9e175865893 : SUCCESS in 6m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82f2bf8d60b64d5d84334e8f832e6b34 : FAILURE in 48m 57s","accounts_in_message":[],"_revision_number":6},{"id":"184f9c0a295278f40b56a1f1cb2deec838821a40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-27 13:08:40.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"796af2e5e33663021ba6e8cff7d895e18f8952fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 14:58:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d3e0ba2a939437fb54bf58ff8476994\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9bea53fbf5db4b41921daf0a05a4c852 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6431262b3d6046b2b52ea8b9300f22d6 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1d24e5bc23473c912865774e7770b2 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa25c478cd9496fb9b44f36172c4d9c : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb30e83baf1c4f1b850e55676eebcf75 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/880c4e26a07f4906802822ff032ef6b6 : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeae5879551f4f778332b0741470c408 : SUCCESS in 1h 17m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ad477a47bbd46b0a381b851c7153525 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28644d253249464db4575716d2e6eadb : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9f769960ef041b199535426fc2f2fdc : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fe7786840b5a454e84c8b6ac713e2846 : SUCCESS in 1h 12m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/40e5084391474930a5e78f90542e43b9 : SUCCESS in 1h 48m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6820e0de042142a29063ebc6fec5648c : SUCCESS in 34m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/17a2d5b101b94944972dfdb3fbe8792b : SUCCESS in 6m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52eca0858e1c44f49dbfb1d0acb32930 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":7},{"id":"a4bfc7f1c47a64dbb776f6056f4660290ba48496","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-27 07:04:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cfc489f9f782da441509844d167b87bac844f5fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-27 13:55:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5489b4ae4e184cdee62b62a8bb35f3f3566825cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-01 12:57:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dda4540c3cc9c0c6a5d231f9a97d9bef31721215","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-22 08:32:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8ea03d6eddc78aaa173b99c49a380a8f38e8270","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-22 08:32:28.000000000","message":"Patch Set 7: Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"3784a50f45a1bbd7a6cbed1b5dfac4661c13b5ea","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-08-22 08:38:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"78849eb2f7d9feb51178eaf212a7304d1c7533ee","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-08-22 08:46:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eac838dc884c71daedcd6b7931ba6020144b20c9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-22 09:01:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"045ce71ae279e68f10d48142106d9cc676bebbeb","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-08-22 09:14:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e479ae88a98f8aeaf54700fcbf32b49e7b1667eb","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 09:42:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9d1e8afd95fff6b63bcc5bc5ba1c83120338a179","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-14 10:06:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3e14b71d396cca9c446b54c0751970fdfa7dbea2","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 10:15:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8ce7aa8baac2ae1f96f0ded3b1fa47b6eb8c36c4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-14 15:01:35.000000000","message":"Patch Set 7: -Code-Review","accounts_in_message":[],"_revision_number":7},{"id":"a774cf153989a4a05caa9f78a836b7b95309ce9e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 15:22:44.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e29c3e91e60255803d5dca87dff43239797e4730","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 15:24:50.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f3c59c2efd55911ef4bce365f1052a90aec55ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 15:33:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1a53e567e0335b2befa6b43125f12c354a4c7479","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 15:35:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"85b18bc52c2894a441b501f6a5ce4e1207782283","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 15:44:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c2ca8455f1f8a589bd3e616449458bd21ed35106","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 15:44:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8ad5f5f9577609171e74ffb4d4af649346ec8422","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 15:45:31.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b61170dda0219650ee4e302fcc0019495847a766","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 15:46:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8adbbe87b0b5515af7213b870ec42f467ec6de16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 18:07:49.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/548b1da0dbe64b30860491f619849a2e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f4d8474906b94ef787d7bf09c8a48bec : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4163374ebab4255b71f811e8c2a5eb4 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0ce3584db4c47e1a07f50bb115cebe8 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bf61fcba8c4fcf82640538bba6c870 : FAILURE in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d96eef9613246cf86000c78c9b42b05 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9e06a0f06534b15883b3d865790aae1 : SUCCESS in 3m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70c6d67b57c54ccc8dff639408739807 : SUCCESS in 2h 03m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418672757a1f44c7b9a69ea581ea0243 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2e2ac2e3006142e7b50ab93a0e119837 : SUCCESS in 1h 52m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b108593ad90648ad8c42b2a474c3f85b : SUCCESS in 1h 54m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/365c69057c8b45a2aa30ebbf28267aaf : FAILURE in 2h 21m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab51c6b343a74f6b951e23340c5fff6a : SUCCESS in 40m 39s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25d25c67316f41078a5b0fe361580fe0 : SUCCESS in 1h 48m 23s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0308c5d0d9b846b9afb847a41c656283 : SUCCESS in 2h 18m 59s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/2fc5b803bd8c49c2bc7034d98ac9ea22 : SUCCESS in 19m 34s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a0ba31a2d62a409eb90833d4a923f1dc : SUCCESS in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7463304f81a24808aec591fa186b8ec2 : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":9},{"id":"707bf5b36daed01dfeff956d24ae9fd21a771c48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-14 21:11:19.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"a7ec2978472bc88c860522f9fcf73059ecac8b8d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 21:31:54.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"5545e3c73a4969a3d3d3c94bfdd303712e4807c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 23:13:38.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0a7f4c64bef4c43be505fce0d3f504b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9e29306cbace42e2a90fc40c998a30bd : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c6da12b45d414988852003adb37e5a : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197246a0e3a043f0811f2a4062bf0409 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694eb2e463384286a9354361761feac7 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3305a3a3ba4e8d92e599005eb35f02 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/175f6b5108c64bb1ab3b4763ebec945c : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/661f745d91094f6ea31acf4ef3a76c40 : SUCCESS in 55m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6a522aabad44a65be68a05086de1535 : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a197d95dbdf4468c8a5620094651a7be : FAILURE in 1h 33m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a66d3f37252146b09d7d73167ee1d04e : SUCCESS in 2h 00m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c842e395e5940b99dd47e1bce49f8e5 : SUCCESS in 1h 59m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/319b852842414c808002a78929d1993e : SUCCESS in 35m 23s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7220c44cf4bf41bd91eea5184bd4046a : SUCCESS in 1h 24m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1b4044762b6422fa4b17bd5a3d733fe : SUCCESS in 1h 38m 58s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/e9f446d422c4457d83f1aece6c678011 : SUCCESS in 21m 41s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0d2e68499c1c422a8fa38a31c6cd7804 : SUCCESS in 4m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/479b56a6ef26477eae22740605fb6cd8 : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":10},{"id":"af376555ab0b6df629440081ee42b4d523ec5c3b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-15 07:12:37.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ea67680b244ba497cdaa2b5497eaaa8b445911e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 09:24:33.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c897766e0f124f448c6fac4736399b36\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/da229c3af4304ccfa412062cd3dd2d67 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69f532f0d7248b2be9f714e69cf3550 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d61132ae98aa4c2993f929cd1b8fb4ae : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab87477e20f4ba2a65185716c5b6cb0 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51d3d00aaca54bdcae8f987b0d8ec55b : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3aa7d722ee55412ba811f9e0683387a7 : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc75338bc119403194d11105d9931e68 : SUCCESS in 1h 40m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0721534b487d4caf8ca5d05d205c92f1 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/33192942983245108c92b25f86286f28 : SUCCESS in 1h 49m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6826fc9a28b4fca9d5c8bcd6d62b5b2 : SUCCESS in 1h 47m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/019215f0a8f6445d8c0bcabfc51ae3ee : SUCCESS in 1h 59m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a6b161c57454fb99ca03edd323c8781 : SUCCESS in 38m 28s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9edeaee67e99439eb9e4aef8f7081367 : SUCCESS in 1h 38m 30s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69828c3078d9482dbf9c3779457c9f0a : SUCCESS in 1h 59m 36s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/53031633ea1d442593a3bc7271fb66b4 : SUCCESS in 20m 24s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ea60d727f00b4f38b7c2221a2d744571 : SUCCESS in 5m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b9c6f9446db04ce09c3f1ba6337ca0c6 : FAILURE in 56m 42s","accounts_in_message":[],"_revision_number":10},{"id":"410c965644be83f57d8f82105a161cb45416cfcc","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-15 22:54:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"423b000b1eabcee2cea918a0658414b62b70ecca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 01:17:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76455e1689b4f649bd5afca659b0172\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d5d4ece387fc494fa953b4ddee398088 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b192b6f4aa40b88903dd4412a9a39b : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a548653b38c54e91b94f5890646543d0 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ae508db04240e2b0b0f19486a9941c : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3d7053b178f4a93b299f1adf945594f : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae76671829ca4301afcb83b428934eea : SUCCESS in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e81a193ac2fd48d4b1a767a2a899b09f : SUCCESS in 1h 00m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/813b5246aa9a46fe98179f541e6ee6b4 : SUCCESS in 4m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cc28eceba0984ae8bae1bf360b77349d : SUCCESS in 1h 33m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2647f180866a46e0a3f167c3b149f007 : SUCCESS in 2h 07m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/436dce7b134a4e5f95f826a0b62e2227 : SUCCESS in 1h 44m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/29894600ba504bf2a8c5754b9991cdfc : SUCCESS in 45m 28s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c630a2022cb04a1597fff44c650638e2 : SUCCESS in 1h 53m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb372b0f27584df1a19e6a206afc5380 : SUCCESS in 1h 49m 41s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/2d317b5d234343b68a179f3f909daef6 : SUCCESS in 21m 06s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a06ce778144147dfaba8e5c0515b2733 : SUCCESS in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df334c0495b449609640001e142fb349 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":10},{"id":"8fbcf32bb891f44d2dfacbf59c17e8ce3ff04318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 01:18:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"aae20a236161735ccd02f98dc034a1e79d811e49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 03:11:58.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ce0056bf839483f8f7b8eb2e4a2296f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d047dd95bcd3470e85654692f1966b3e : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049ef6d3441c44fe81e9f5c2313a12db : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e7bc2bcbb74c10aa2cb0564db3a292 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af9759bf09824fc0b4eff6dd24260a5c : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f5dc055db714310b1b172376db1cfd8 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78d88567d1c04021a71b3e0cf6631c17 : SUCCESS in 1h 53m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b4c6105b2c4301b93d239eb4379116 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba5818ef32ba44c7b42b89363c985f14 : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a18ec11f5c9d47c6b0126fbd33368b64 : SUCCESS in 1h 52m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e39ce81c30d4e488ee41ef1f0f167c0 : SUCCESS in 1h 51m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9b47b4bab2144d0ea76ce057422a5524 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":10},{"id":"c02deccd9029565cfb862e16d8897c16c040dc43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 03:11:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"104926eba22194a8c22fdc53fa1bf7b1ff520186","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 03:13:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9b3a70fcdfb42eeab98947f39f96545\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3b60a876047c468682215fe588aef6fe : SUCCESS in 49s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b426003ca24553984c6bdcb3a5aa44 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47045bbea8ee4107a80455ef0d3abb6b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"ef72d4cd6e2e74a0452dcea916613357c3627a22","revisions":{"257c152437d80a9f09743b18ff24319e22b1a229":{"kind":"REWORK","_number":1,"created":"2023-12-19 12:10:17.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/1"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:10:14.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/257c152437d80a9f09743b18ff24319e22b1a229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/257c152437d80a9f09743b18ff24319e22b1a229"}]},"branch":"refs/heads/master"},"48ac3b36d86d95a3648cfc23fe535c32d2e69c35":{"kind":"REWORK","_number":2,"created":"2024-01-08 17:56:11.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/2"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-01-08 17:55:25.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48ac3b36d86d95a3648cfc23fe535c32d2e69c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48ac3b36d86d95a3648cfc23fe535c32d2e69c35"}]},"branch":"refs/heads/master"},"a761d6708ca1daa236547c1e0735670205c27218":{"kind":"REWORK","_number":3,"created":"2024-01-09 09:13:26.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/3"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-01-09 09:13:05.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a761d6708ca1daa236547c1e0735670205c27218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a761d6708ca1daa236547c1e0735670205c27218"}]},"branch":"refs/heads/master"},"a7bf94a537be18a1c6301b41e1c7c771efb8ce25":{"kind":"REWORK","_number":4,"created":"2024-01-09 15:58:33.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/4"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-01-09 15:58:21.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a7bf94a537be18a1c6301b41e1c7c771efb8ce25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a7bf94a537be18a1c6301b41e1c7c771efb8ce25"}]},"branch":"refs/heads/master"},"91f6051b2302d9e2783ddcb2517df39dd2e2953b":{"kind":"REWORK","_number":5,"created":"2024-02-26 15:27:14.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/5"}}},"commit":{"parents":[{"commit":"d131b53403b4dc4c1e4730cb723e75b8668b894d","subject":"Merge \"Add extension \"subnet-external-network\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d131b53403b4dc4c1e4730cb723e75b8668b894d"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-26 15:27:11.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/91f6051b2302d9e2783ddcb2517df39dd2e2953b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/91f6051b2302d9e2783ddcb2517df39dd2e2953b"}]},"branch":"refs/heads/master"},"7e76067f539bf3a981c9b20b9e9a38a524d587c8":{"kind":"REWORK","_number":6,"created":"2024-02-26 17:19:13.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/6"}}},"commit":{"parents":[{"commit":"d131b53403b4dc4c1e4730cb723e75b8668b894d","subject":"Merge \"Add extension \"subnet-external-network\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d131b53403b4dc4c1e4730cb723e75b8668b894d"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-26 17:19:02.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e76067f539bf3a981c9b20b9e9a38a524d587c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e76067f539bf3a981c9b20b9e9a38a524d587c8"}]},"branch":"refs/heads/master"},"258a60a81c0b3568a09ebe2eea775e51e1394e3d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-27 13:08:40.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/7"}}},"commit":{"parents":[{"commit":"f0724e9cf1dd761bd14abfcd8b7e16369387c23a","subject":"Skip the enforcer undefined rule check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f0724e9cf1dd761bd14abfcd8b7e16369387c23a"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-27 13:08:32.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/258a60a81c0b3568a09ebe2eea775e51e1394e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/258a60a81c0b3568a09ebe2eea775e51e1394e3d"}]},"branch":"refs/heads/master"},"7358a2fe5e750f4621b574e8babe17a9ae8b02bf":{"kind":"REWORK","_number":8,"created":"2025-01-14 15:33:34.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/8"}}},"commit":{"parents":[{"commit":"a3d1b7629f192392ee930ae8f05f58a774dac3b1","subject":"Merge \"Fix network-ip-availability API samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3d1b7629f192392ee930ae8f05f58a774dac3b1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2025-01-14 15:33:29.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7358a2fe5e750f4621b574e8babe17a9ae8b02bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7358a2fe5e750f4621b574e8babe17a9ae8b02bf"}]},"branch":"refs/heads/master"},"488fcb06d451b4fef966b8cd703e528cdaf33939":{"kind":"REWORK","_number":9,"created":"2025-01-14 15:44:21.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/9"}}},"commit":{"parents":[{"commit":"a3d1b7629f192392ee930ae8f05f58a774dac3b1","subject":"Merge \"Fix network-ip-availability API samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3d1b7629f192392ee930ae8f05f58a774dac3b1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2025-01-14 15:43:51.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/488fcb06d451b4fef966b8cd703e528cdaf33939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/488fcb06d451b4fef966b8cd703e528cdaf33939"}]},"branch":"refs/heads/master"},"ef72d4cd6e2e74a0452dcea916613357c3627a22":{"kind":"REWORK","_number":10,"created":"2025-01-14 21:11:19.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/71/903971/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/903971/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/903971/10"}}},"commit":{"parents":[{"commit":"a3d1b7629f192392ee930ae8f05f58a774dac3b1","subject":"Merge \"Fix network-ip-availability API samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3d1b7629f192392ee930ae8f05f58a774dac3b1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-12-19 12:06:02.000000000","tz":60},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2025-01-14 21:11:07.000000000","tz":60},"subject":"vpnaas: add support for AES CTR","message":"vpnaas: add support for AES CTR\n\nAdditional choices for encryption algorithms in vpnaas policies\nfor AES Counter Mode (AES-CTR).\n\nRelated-Bug: #1938284\nChange-Id: Icda2da71135065a1192954f33943b9ee225c5cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ef72d4cd6e2e74a0452dcea916613357c3627a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ef72d4cd6e2e74a0452dcea916613357c3627a22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
