)]}'
{"id":"openstack%2Fneutron-lib~905299","triplet_id":"openstack%2Fneutron-lib~master~I32ce8b3501a01552fed67ba24602e891c3c01cd4","project":"openstack/neutron-lib","branch":"master","topic":"bulk-create-sg-rules","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2024-01-11 19:40:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I32ce8b3501a01552fed67ba24602e891c3c01cd4","subject":"Add api-ref about SG rule bulk create","status":"MERGED","created":"2024-01-11 11:54:50.000000000","updated":"2024-01-11 19:41:03.000000000","submitted":"2024-01-11 19:40:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905299-bulk-create-sg-rules","meta_rev_id":"21dbefaf1f0eea624dd882b9a8516fc96c4cb5a7","_number":905299,"virtual_id_number":905299,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-11 19:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-11 12:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-11 17:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-11 17:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-11 12:27:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-01-11 14:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 17:17:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"279f2d4ee5a074492ddce37c3d6cda20f5033b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2024-01-11 11:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1f6f290f7e8e9223f58129565402b67788a73c7","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2024-01-11 11:55:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdae2e635b9104023cdb4b46456f1bda5909b004","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-11 12:27:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bc13800a3233a45b788d7ac02224165e67dd752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 14:07:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43e894e2a904c0db5b6a21d80ac967a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b49110a04014952af0e3d13a415dc33 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817b34ce08dd42ee8dd53bc54275e68e : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4bb6af30034a5393019fd9ce86926a : SUCCESS in 5m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/057a08201f904834b0b9f083a3b5bfb3 : SUCCESS in 1h 38m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc854fdc36984d1e9431c7b8f443d33c : SUCCESS in 2h 10m 11s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fb6ee665e3ac46bb8d5d4640e163c506 : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6933352ed094bf59717f88cd9dc1521 : SUCCESS in 1h 59m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c80e87f972b24baca7feb3e69672ca3c : SUCCESS in 40m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d4051b1912746caab1876b227bd9385 : FAILURE in 46m 40s","accounts_in_message":[],"_revision_number":1},{"id":"050f9c23afca991f30806a471ab24af7df05e377","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2024-01-11 14:54:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53183c80d9217dc0e173c0e19221759aaaeeb4dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:48:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4b9269dc3a4d658a779253b0f63357\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/244393fbc3ac4aada0c06026f06d629d : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03779d570c2d43af985c278db03864d2 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d7597a7b3074f2aac5670a418472cbb : SUCCESS in 5m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/146193045d32421b9826099db989854b : SUCCESS in 1h 52m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49af11ba84224d3bb25b7d610e45b280 : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c76ed3e730cf46d6ab039e1a40f78510 : SUCCESS in 1h 17m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6e4006a6ccd04e859b91402f57d89607 : SUCCESS in 1h 24m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d791bd344e6141719a9733aef20f6458 : SUCCESS in 31m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37d4318a7da5458aa3956afb7fb0c5c8 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8c75fc68fca1a6c51f54df1a15d1b3fcb9bc0b86","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-11 17:17:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3d7c8cf5cc9787a862e0b5f166d587f67e63dc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:18:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"126da6db3fe22567d4f2b9bdfd9631b2bebb1603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 19:39:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d22a5bd2287a456dadd04dc887c09219\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2d4ad59755ac4710b6ed6315ae511b77 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53bebdc6a634416889f248f5c03f4ef4 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1edbca15e848446389e9857de5caf8dc : SUCCESS in 4m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/107e4ac0d48a40b6b07953e4d838653d : SUCCESS in 2h 18m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2fc0773237ad4be9af16638182af190a : SUCCESS in 2h 10m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/136fab46f4a5401688487c963341942d : SUCCESS in 1h 26m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ab2f621c511146818deb2a46718eed56 : SUCCESS in 1h 38m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e4e39f5a0204b52b46b5aa85e28c3f3 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0f536962272f550227ba15d63cf2f10f25558a6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 19:40:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21dbefaf1f0eea624dd882b9a8516fc96c4cb5a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 19:41:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210915f3d59943538814de734581f551\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e0d87d3c47004d718ab7a322ed63c6d1 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e29064fde64497bd45baed839ba3ec : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2aca167054181675837ab0e1465d8e683cbad0b","revisions":{"b2aca167054181675837ab0e1465d8e683cbad0b":{"kind":"REWORK","_number":1,"created":"2024-01-11 11:54:50.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/99/905299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/99/905299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/905299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/905299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/905299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/99/905299/1"}}},"commit":{"parents":[{"commit":"9e3a3a608670d2d7bc0ae98fd39551920e563efe","subject":"Merge \"Update default for BFD/ECMP router extra attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e3a3a608670d2d7bc0ae98fd39551920e563efe"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2024-01-11 11:51:56.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2024-01-11 11:54:49.000000000","tz":60},"subject":"Add api-ref about SG rule bulk create","message":"Add api-ref about SG rule bulk create\n\nNeutron support bulk create of security group rules, but that\u0027s\nundocumented in the api-ref. This commit fixes that by adding the docs\nand examples of these calls.\n\nChange-Id: I32ce8b3501a01552fed67ba24602e891c3c01cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2aca167054181675837ab0e1465d8e683cbad0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2aca167054181675837ab0e1465d8e683cbad0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
