)]}'
{"id":"openstack%2Fneutron-lib~905418","triplet_id":"openstack%2Fneutron-lib~master~I29e2375b68dc01f6e042717a6da3861cc0bd8f88","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-12 18:01:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29e2375b68dc01f6e042717a6da3861cc0bd8f88","subject":"Cleanup setup.py and requirements","status":"MERGED","created":"2024-01-12 08:46:51.000000000","updated":"2024-01-12 18:03:55.000000000","submitted":"2024-01-12 18:01:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905418","meta_rev_id":"14f9cc25be77d2205e90e0cdc6a9ee5b4d1db49d","_number":905418,"virtual_id_number":905418,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 18:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-01-12 13:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-01-12 15:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-01-12 15:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 11:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 13:17:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-01-12 15:31:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f24feab2b2dca67c0ff78efc7a38c32a9ebf7f07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-12 08:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f24586331766e028bae1d4709fdc27f27029a8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 11:53:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe7b025a3904f888d2fe1c8466b3a4d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b5e402b7a69404caf93b3d8ca82bcad : SUCCESS in 4m 58s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fd10b333acf41af914c300dce00bfaa : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca096deef5ee439b80fc64034eaa263a : SUCCESS in 1m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2807c86c3a4ade92df208c5f2e14f8 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2093eca00034a2699a31a429e083663 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66959c9a34a42efb6776c3adf8e2c51 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2de353eabf164e66af4b632cedf10849 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/586a9461b7554371aaaadbaaa82df49b : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0f947a16c4a451ebe87d23e31ecb8a2 : SUCCESS in 1h 36m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a0a2464a944650abdc5f69fb62ffb7 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b5f0c064bfba4b94b3568869e7701370 : SUCCESS in 1h 56m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2f2898377a4411a8d8576b86a44253e : SUCCESS in 2h 08m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ac05e134688140aa85aac33080fd1808 : SUCCESS in 1h 55m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b932321853f54ba0a44b74c7fb69b464 : SUCCESS in 2h 34m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3392908ae7ad437fabcacdf9c2956fa4 : SUCCESS in 35m 01s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e68e24f6b8da4168942cf7b65610f4b3 : SUCCESS in 6m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62f25603b50642a29e46777618f5227f : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c50d97a4cdf47bb9ce49d86b98e4dc3fb76e73ce","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-12 13:17:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb3bd9365cf72ddb76bc1ab08b2b6ee04a14752b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-12 15:31:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e371313c88de4d294c66677658e0cef6c0c792d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 15:31:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fc42e80cb33972e8df79814dd8f55cb510ed6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:01:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dabd6c893b54b7f90da0c031b292149\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5f6691c9b3604b56ad6ebbc019c30fec : SUCCESS in 3m 55s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4a4bffb5f0b466e9a22490ce9cf1a51 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ca8f08bdcf9465786caa8486eb625cf : SUCCESS in 1m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c89ea697ce4050a2b50fe5ea9956aa : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d23eab2f294bcca4c344d834a94412 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c09e1ce944d4e19a73ce23f0d2e7f41 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f8b8f841dfa4640946893424470f18a : SUCCESS in 5m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e961254c36bf4ed892c25d55ecaccbc4 : SUCCESS in 1h 40m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94bdbeed5b8a439f80aeaae49766a7f5 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1254d5c0f83049de80f07e19c1c1d5ee : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c2b7ab4922b41f59de206da376e63e0 : SUCCESS in 1h 57m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b339abfa862748fc8021a2db41766cca : SUCCESS in 2h 25m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3dd9622113234675855546b7e0464bc8 : SUCCESS in 1h 44m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/018cc3217bf649dc9bd50fc037160396 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e930522d426633aaceb0908e773bc73e381ee30d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:01:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14f9cc25be77d2205e90e0cdc6a9ee5b4d1db49d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b15a42686f947acaf7210d91dd5e9ae\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/75101397d71f44b39d92953b88c45a45 : SUCCESS in 1m 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9066a88b700d4e35aacaeb8ba97ab634 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/503b740c96564e7f9298e71e16a33d7e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27689c9ce535fa87615ca5632afa1900b9df5393","revisions":{"27689c9ce535fa87615ca5632afa1900b9df5393":{"kind":"REWORK","_number":1,"created":"2024-01-12 08:46:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/905418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/905418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/905418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/905418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/905418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/905418/1"}}},"commit":{"parents":[{"commit":"b2aca167054181675837ab0e1465d8e683cbad0b","subject":"Add api-ref about SG rule bulk create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2aca167054181675837ab0e1465d8e683cbad0b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 08:46:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 08:46:28.000000000","tz":540},"subject":"Cleanup setup.py and requirements","message":"Cleanup setup.py and requirements\n\n- Python 2 is no longer supported\n- setup.py is no longer managed by the global tooling\n- Recent pip does not require appropriate order in requirement files\n\nChange-Id: I29e2375b68dc01f6e042717a6da3861cc0bd8f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27689c9ce535fa87615ca5632afa1900b9df5393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27689c9ce535fa87615ca5632afa1900b9df5393"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
