)]}'
{"id":"openstack%2Fneutron-lib~907949","triplet_id":"openstack%2Fneutron-lib~master~I75a9c30f1e8031d40a548df345b02fbe0bc47706","project":"openstack/neutron-lib","branch":"master","topic":"bug/2051831","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-02-23 16:26:12.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2024-02-23 16:26:12.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-23 16:26:12.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-02-23 14:30:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-02-12 19:44:28.000000000","reason":"removed on reply"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2024-02-23 16:26:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I75a9c30f1e8031d40a548df345b02fbe0bc47706","subject":"Add extension \"subnet-external-network\"","status":"MERGED","created":"2024-02-06 06:35:50.000000000","updated":"2024-02-23 16:27:12.000000000","submitted":"2024-02-23 16:26:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"907949-bug/2051831","meta_rev_id":"04afe58cf1706dce7da7fb30fb32e57f63d428de","_number":907949,"virtual_id_number":907949,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-23 16:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-23 14:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-02-12 19:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-23 14:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 08:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 15:03:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-02-06 15:46:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2024-02-06 15:46:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-02-06 15:46:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-02-06 15:46:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2024-02-21 12:07:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f9abb1e34fe7efdf5dc90efb23ebf5a7ff0972bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-06 06:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0de599ba15a234b29a08eb595aa233d312505cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-06 06:35:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"250af7903d7449a465f52c09a5c1cddb19b3a15a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 08:37:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6b0857a64374285ac60dd86b499f38e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b80a5c40404444a09b3f4332d7f52c : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b27efc2b8404003bc7feff773010d6b : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f35d7eb6134f959577e9fb38dc102c : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ea924734c242c09dd82ddc25fff797 : FAILURE in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a0c919c800b44e6ba1644b10d2a46cb : FAILURE in 4m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7545e9f4dab742a8ac8e4ffc153e3b83 : SUCCESS in 1h 32m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4f741ed345a94fd4ac288ffc0300b676 : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28cb6ce3f35343c6ad74ce782f753419 : SUCCESS in 1h 43m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b236abd93144453e9635736db03e9e6c : SUCCESS in 1h 45m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2a6b980c259e48f1856a19f6aae67e80 : SUCCESS in 1h 59m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b58f702fb5cc4ae39d8a848a4fc0d908 : SUCCESS in 50m 44s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9a98e40a730145f381a64aa7053aa818 : FAILURE in 5m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db1f2d2dd7f746c7b2b90036cdf76692 : SUCCESS in 48m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9bfbe0175b64853c6bdd37a4975b8282df2f1826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-06 15:39:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"66ce80086fa69b5b1806a128985379bf1536afb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:11:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a71cd8b5e40a42359945d22f8b065934\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b8ca36cddb445d79c281e9bac47f78c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63bca44648c49dd948cadd567a003a8 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e4a2583ae8345e98f603fc0f5ba94c3 : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936785aa38e647d48947d796de8f341b : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6486326fc2f54d8c9776fbf881fe228e : FAILURE in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c72a2fc4fe344ece818421fce5ebc127 : FAILURE in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a7131ebccfc491894d9a3a1e65f5521 : SUCCESS in 1h 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfe8e5102a324e82a478184ee3340e49 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef6a6d396e5b41959955bb05f6c8033d : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3e93c0286cd4228a7de81bca7ce96a0 : SUCCESS in 1h 30m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a411795c3b8f4ac1885b34adf1e7e8e0 : SUCCESS in 1h 11m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6d7e8a363b7b4fe69cc80786f8615686 : SUCCESS in 1h 22m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/abdc189d45614708b273d12f0ab7804e : SUCCESS in 39m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c3b83829e22142f6bca33f6f4a96fd12 : FAILURE in 6m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c02cede3920c42f7ac35fe4a76ee2ec8 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":2},{"id":"478449accd877ea8573f8ded2f1d6575e8b79df1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-06 17:59:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bcc205a35a062852eba1dacc0db45d9e26bbfae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-07 07:49:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f48a1dc7d86f13054f22e38bf6cfaae87c18915a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-07 07:50:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eab8fa24db84a462b8d6324602e92c8737c4c945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-07 07:51:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"20900cb459e0c118ce95db160024ccfcd892b150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 09:45:21.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5072c00be52042469aacd4e55059f190\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/47747bd6d07e434c9a03c5a8da435ea3 : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483c47d800e74735bf5c7e6b21cc9ff1 : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06a641522e9449929814b2276b6be801 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35eedc1e43eb4cb6923ed91927781b1e : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27bdba06514042b9bc192ba2a332d89e : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc44fe36d0ee455aa5cf957c343b0db2 : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f642d4eec65d4c31aa3b08ba41b72d42 : SUCCESS in 1h 02m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4871eb74bc2f48e3a34bfe7fc92c5cc4 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4053babf4db940f4b1c184b498c2664e : SUCCESS in 1h 27m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf590c6649214f2cb43ad7fd17b57974 : SUCCESS in 1h 51m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/500acd59581e4e65a7b918cf2f591166 : SUCCESS in 1h 33m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4630cd1a05944aa8a82b7a4b15c01c8d : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/388d628182d94862874e7ceeab9b6960 : SUCCESS in 34m 15s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/519445c2a05b49bb9bf00ea8bc971953 : SUCCESS in 6m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c642662313f4490ad33eba289d1be9c : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":5},{"id":"02db0c1a06ce0c9478aa97a22a26a58bfadd02ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-07 16:09:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0f133fed15ff204d68bc59ccef1a59308937fcea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-07 16:16:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d94f50d5137f22a3ae35ef79f2a764f90a3c71ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 18:07:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf49be9419dc49fabe663bd43c5ad91a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50ac270ae0fe4a06b826610b03598de6 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2c7496700a4a71a8414b55839222f0 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f82bc0bdcee04ef281e38326b0cb538b : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05429a749c6f4b7d892c5d5fdc739f92 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79d9c2b531f4ed2ae3c39d33f2febe8 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdc1d62655204fcfad37a6c9ab0d10ae : SUCCESS in 5m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdb5ec921b8042259485bf4bae7ce6e4 : SUCCESS in 1h 19m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac531b44f52044e49faecad3e8692c2e : SUCCESS in 3m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5f66e48f81eb4a11bf5fb2360a062324 : SUCCESS in 1h 17m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f434af0ea0e14731a0ae954583df6505 : SUCCESS in 1h 55m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7a600d653bd246c6bd0a19f7fbdaaa95 : SUCCESS in 1h 27m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a4fdbd7685d14421af8bee4b631604d9 : SUCCESS in 1h 35m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/93928a48b42f475a8d8f9bfd26ec228f : SUCCESS in 32m 25s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/199e993eb1bf40e7a6dc4ccf8b904c9a : SUCCESS in 7m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e33fce06e1b24a07ae0d3386f4900150 : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":6},{"id":"1c6e84f4f3111de206e87fb9361d689ef34eecf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-08 07:46:34.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"41f54d9ea7b948c40bb4ac21fc837465ad7cb88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 09:33:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/535b7fdec23043ce834f7cc324b63520\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd88672c2ba74fc0bb67aa761d5f62ac : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1261f96fa96542609895a1dfe033335a : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b74a2c044a04e91b367635de22d9c80 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77774ce2c4aa4510be8018a5177afd4d : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5531453fe9404e19b8af2de64bf53381 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/511503a5d27c48e68d88afa6f20cf594 : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27fc279e40b7467ea27d512a808d6e20 : SUCCESS in 1h 27m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/587490f916774ea082e8d4960718299f : SUCCESS in 3m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c3ba1a60761346819282d1412bff4d46 : SUCCESS in 1h 20m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca84449ff94244d4b5af08bdfa58a6bf : SUCCESS in 1h 26m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8735d21d28fa4b4385eee759b13af5e5 : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1b8927eab17e450684bf2ed4ce7b442b : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec0377405d4e431899616f7f6a817d28 : SUCCESS in 32m 11s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5439cb8cec3c417ab917fdf2826849cd : SUCCESS in 6m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa3d8ad750424d6f8b0673c36227fe3a : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":7},{"id":"fc95b2d881a14c09074b2e50b07a9485c99a5a57","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-08 20:02:21.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"300e68a2c5b74d45789a5cf02a7edad7cf342291","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-09 09:24:00.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"376fe04a7c81f77cdf0969863949a12e4c798fc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-09 09:24:39.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9b2145a8749bce3922e027591ada418f427f28c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 11:19:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00c0504b39724235a3462e2f356d237f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45f32533b3ad42dfa72bf490ad8b453e : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602b1901c8b74da484000017917441e6 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e002cec19584b5e914f54ca6e83955a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f3cfa60a1b74224931e48b9511a554d : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54af9d0524e94b6d8dc5eef690451136 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9accc8941dc4b71acd5f5cff1f5bba6 : SUCCESS in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/238a28c9adea4891833d2311decd3738 : SUCCESS in 1h 29m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee11f87dad9a4cef88b86f3c18b58f20 : SUCCESS in 3m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f2cf6b0342864da29912cd31128ff759 : SUCCESS in 1h 21m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53066bebf28342808823d54617172357 : SUCCESS in 1h 33m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/923bf184f370447b95726e29441d7acb : SUCCESS in 1h 15m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/08dec20951de4d7eaab7b19d684d3edf : SUCCESS in 1h 51m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8fece467981242c2829707aa1934871d : SUCCESS in 39m 33s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8e09563abae4480fbb8fa611b40305d9 : SUCCESS in 6m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1a800e24a5445f6a1309ee8aae5d98e : FAILURE in 44m 53s","accounts_in_message":[],"_revision_number":8},{"id":"7a8dd3e9041723f974cf6b129dae760af330cc87","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-09 12:39:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"af2a40f791348ea6aa66a8f08f5d70dd973c5aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 14:21:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/058ccb29a9e646948e1ae258615985fd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fd023433e8624f85b3323e52f06a9816 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ad45a8559c4b14ac1910ed65f6b10c : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bc6b47a625545ce931972757773208a : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d261fb13d04fb78d70bbfb8ba907c9 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb11d0b2a07141f181ffaf2be7f29b62 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0582e1378e9f44a487bdebc67f5b383f : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec760bda697a48f2a12aabc8a98a5a15 : SUCCESS in 1h 41m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d911ce45b58d43e7b0abdf77bc22d94b : SUCCESS in 4m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/25d367af74794eafbd824121a51aeae2 : SUCCESS in 1h 20m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e11af439a304615a66b3387467f6579 : SUCCESS in 1h 30m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/53ae45acc6934538ace6d0b5edd2666f : SUCCESS in 1h 24m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/257587b1deaa47f98092f927e9cc3083 : SUCCESS in 1h 25m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2880e879651a4e32b2aa392c1a6f6091 : SUCCESS in 45m 27s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dfed4b9d1477475fb627af7cc9c21732 : SUCCESS in 5m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a8da5356cb74af3a1d5932ee3a1431e : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":8},{"id":"bc83233ac37aa359d9a399bec8094792efff838b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-09 23:19:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"22a4b22b0889f21220b6bef2ef82b62103da7111","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-12 08:11:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c24f82dcf50d97aa47830a698886aec818f22c19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-12 19:44:28.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"4eb1613bcd3a50c2c44f8343105f89c8481d81bc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-23 14:30:51.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e40e3cb70b9df11ceeef246751c073138be85dae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-23 14:30:55.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"3e6a569d8e0d8b93411477a2b55fef16dea67ff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:31:26.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"ed9a2eada033724eab1df472a8c1a7e65610cf24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:26:11.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5944d32d3f7d41288629bb9e2ede8e4d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/11ab0532f63b441bb650b5248a52c041 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccf5ce0eb264b88b9744ccd19e3d96a : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5636728ac231408db398d9568a4e4de8 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de6ffbc54883430ebd373dc5a92bc3b0 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c60ddd90383145369acafbe4fa47b217 : SUCCESS in 6m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcbf970fe08f4b36bbfe28f2388007c7 : SUCCESS in 1h 41m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d4f05f016544b47bea3697cb3de48df : SUCCESS in 3m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b01159a19a5b41b1a3ab9c21d699968d : SUCCESS in 1h 52m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e62979d8e6d94dc0be82be63fb7c7d87 : SUCCESS in 1h 49m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/360f1b3fe2174d2b97ff0dde7af1ad2a : SUCCESS in 1h 44m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/71075783a6114ab9af0585423b88cae1 : SUCCESS in 1h 32m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd11e642045740e6833f1df96fcb76a8 : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":8},{"id":"0cc954a6cad5e511ac0c10bfafef6897ffeab094","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:26:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"04afe58cf1706dce7da7fb30fb32e57f63d428de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:27:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be2476da70e74e5bb769c9d21d53a0f6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/06ea156cab1b40fdae2cf372a08ef491 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28400c2ff40e4853ab9a9fd28f8de2e7 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adf41f0f9c96461eb329b593d17830e5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"fc2f383dd698321f01dc8ec56dff44ba4ec66157","revisions":{"43ecbbb9c3387888d541e52dea7737341c462c26":{"kind":"REWORK","_number":1,"created":"2024-02-06 06:35:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/1"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"subject":"WIP \u003d\u003d Add extension \"subnet-external-network\"","message":"WIP \u003d\u003d Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"external_network\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\nRelated-Bug: #2051831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/43ecbbb9c3387888d541e52dea7737341c462c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/43ecbbb9c3387888d541e52dea7737341c462c26"}]},"branch":"refs/heads/master"},"e205973482f75bc1bb0d87b6d341b62669ac37e7":{"kind":"REWORK","_number":2,"created":"2024-02-06 15:39:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/2"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-06 15:39:37.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"external_network\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\nRelated-Bug: #2051831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e205973482f75bc1bb0d87b6d341b62669ac37e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e205973482f75bc1bb0d87b6d341b62669ac37e7"}]},"branch":"refs/heads/master"},"139722c02b3d284d32bf72c70409f1c68e637625":{"kind":"REWORK","_number":3,"created":"2024-02-07 07:49:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/3"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-07 07:49:00.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"router:external\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\nRelated-Bug: #2051831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/139722c02b3d284d32bf72c70409f1c68e637625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/139722c02b3d284d32bf72c70409f1c68e637625"}]},"branch":"refs/heads/master"},"30ecf61456543ec30753dfe7141a502ea2b75263":{"kind":"REWORK","_number":4,"created":"2024-02-07 07:50:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/4"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-07 07:50:35.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"router:external\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\nRelated-Bug: #2051831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/30ecf61456543ec30753dfe7141a502ea2b75263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/30ecf61456543ec30753dfe7141a502ea2b75263"}]},"branch":"refs/heads/master"},"c394c2942cf5cf56b85691ccb7c17ade086a3f26":{"kind":"REWORK","_number":5,"created":"2024-02-07 07:51:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/5"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-07 07:51:43.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"router:external\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\nRelated-Bug: #2051831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c394c2942cf5cf56b85691ccb7c17ade086a3f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c394c2942cf5cf56b85691ccb7c17ade086a3f26"}]},"branch":"refs/heads/master"},"f59d8deefe364681663441ce7c2e58ad8ee506de":{"kind":"REWORK","_number":6,"created":"2024-02-07 16:09:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/6"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-07 16:08:43.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"router:external\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nRelated-Bug: #2051831\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f59d8deefe364681663441ce7c2e58ad8ee506de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f59d8deefe364681663441ce7c2e58ad8ee506de"}]},"branch":"refs/heads/master"},"0d1fb036cdb3fb279cd10fe1fec5915ff06e03e2":{"kind":"REWORK","_number":7,"created":"2024-02-08 07:46:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/7"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 07:46:15.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"router:external\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nRelated-Bug: #2051831\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0d1fb036cdb3fb279cd10fe1fec5915ff06e03e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0d1fb036cdb3fb279cd10fe1fec5915ff06e03e2"}]},"branch":"refs/heads/master"},"fc2f383dd698321f01dc8ec56dff44ba4ec66157":{"kind":"REWORK","_number":8,"created":"2024-02-09 09:24:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/907949/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/907949/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/907949/8"}}},"commit":{"parents":[{"commit":"a5d8a6fd7c420e0580ff9258f873b860577bd0fb","subject":"Merge \"Remove unused reno from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a5d8a6fd7c420e0580ff9258f873b860577bd0fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-02 06:26:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-02-09 09:24:38.000000000","tz":0},"subject":"Add extension \"subnet-external-network\"","message":"Add extension \"subnet-external-network\"\n\nThis extension adds a new field to the \"subnet\" resource:\n\"router:external\". This boolean field, that is False by default,\nrepresents if the subnet belongs to an external network.\n\nRelated-Bug: #2051831\nChange-Id: I75a9c30f1e8031d40a548df345b02fbe0bc47706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fc2f383dd698321f01dc8ec56dff44ba4ec66157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fc2f383dd698321f01dc8ec56dff44ba4ec66157"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
