)]}'
{"id":"openstack%2Fneutron-lib~909039","triplet_id":"openstack%2Fneutron-lib~master~I3669e11839a99e7d1f60b5be16f1097295196628","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-02-16 15:14:19.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}},"hashtags":[],"change_id":"I3669e11839a99e7d1f60b5be16f1097295196628","subject":"rpc: fix some typos","status":"MERGED","created":"2024-02-14 18:33:48.000000000","updated":"2024-02-16 18:38:49.000000000","submitted":"2024-02-16 18:37:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909039","meta_rev_id":"fed713201ab6070b51acd961b7675bbb3b6e76ec","_number":909039,"virtual_id_number":909039,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-16 18:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-16 07:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-16 12:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-16 12:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 20:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 07:05:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-16 12:27:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f79fa5598dc3793179407df73585d645abe8ba8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-14 18:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"522103b750a0dd9fe11ca0527323be3efd4d78ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 20:13:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e375c7a5a4dc440c9f6c8b954d941b06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378992149b574751b27d1b31e27d24c6 : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff0d2abeb5be4e5899c8d5c5d83c9aaa : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e090dcd8a3f248a082bcf506d46b90a2 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e96a9fb0bc7f4b6a8c279a19365fc390 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c87777b7fec4537a36c334e22c2ab62 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dbc9ee7feb741a88f9f71108e247c18 : SUCCESS in 1h 33m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/10270d90583c4bc8b0a5e0e0f66654e2 : SUCCESS in 1h 26m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4d9e0d0679de400aafe3d4f4432e23bc : SUCCESS in 1h 34m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0a1c220b8b3b4084b15a067fc6232c53 : SUCCESS in 1h 12m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dda914b84c1d4071bcd093b82341a434 : SUCCESS in 1h 31m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f318cf691f4445ab989d1aff12f6ce76 : SUCCESS in 36m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/03af51bd8e714b758bd8385d893ed557 : SUCCESS in 5m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d45ecf165a12481cae2b593ca5612a46 : FAILURE in 34m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9fea752b2a915e717c19a385056f675e2de3487f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-16 07:05:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"824ede70a2e0ae7879baf82fb324c53bc0636d57","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-16 07:05:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbf4eff41919a68f64cdedf73f3cf0cfa77c0474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 08:46:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deed096570cc41e1945dc3f1572b6273\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b48e0ee5c714f48ac0064aec0447979 : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0508739244d4d5c8b7e899a6f16f224 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28411d72374348868b3d2f7d41c9292c : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2a5c2dd5e446f8957bc6775639d8eb : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d72d6806b41c41c792e9dc297b7825ba : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76ce5e33a8bb470b82e26f15750efbbd : SUCCESS in 1h 14m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/374efc674cd245a784a8d502b762619d : SUCCESS in 1h 12m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/875658ddf1734a4e8cd1c5ce4e9470cf : SUCCESS in 1h 26m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/14ba16b8d1844ec4958dc14a1d0ef57d : SUCCESS in 1h 19m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8027861d6d0e496c821a1289848bb324 : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a7e1bc3f6e04bbeb5bd3949194b388f : SUCCESS in 46m 55s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9e2c4344424a4c929c214594538233aa : SUCCESS in 5m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4227ed1accfa4ef89e452bd439d3b950 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"654ad8bef57f9e9e36158da118a93c1819fbb43a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-16 12:27:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78e24c0273a992bafa76ec423acb4d6851ca8d73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 12:28:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c649dd310345fcddd36c6d1e5ac5470ee0a93267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:31:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/884257d20cab4596982377cd113dc84b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4055b4317e24f469233f150e9e6fe8a : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2121f7e129ee4f12b43570f7ad4b3677 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fcfc9c16443407c8aa40a0d5f010501 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d33d5986813644b8996162b9eb8d0ad5 : SUCCESS in 3m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/379da07b7800454b89f4eb13707c1665 : FAILURE in 1h 09m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/36e9b5bae175474989357bb6e8c971a3 : SUCCESS in 2h 00m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7579bf878f134a63b07a749b81f34163 : SUCCESS in 1h 51m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21a42a21f87745e79cb8730f64235d4a : SUCCESS in 2h 01m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/341c5145866a4547b9a9e7a61906f572 : SUCCESS in 1h 39m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8edf8b01c6c74c99bb55918a834c85ad : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ecc29861ba28f4c5a3e2a14807b180f4829ec614","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-16 15:14:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6b9dd37860f741b9051260c4ad3f196fc6ff5c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:52:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f15cdafceee4266b9fd302da5c587b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93f3025bd494e908105b9b5c6cf506a : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87dc5f2e725c4ea8ad27fab70278ffe1 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16b697779ed4345b682c9105d1b7151 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b606881a7933432c9d197836572dbc3b : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c627a8611d6486388b19dee039e0135 : SUCCESS in 3m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/606395ef56fa4fb2950a21a9075651cc : SUCCESS in 1h 28m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99de2f6177bd488fbb303090ca7e6fab : SUCCESS in 1h 36m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d20364be8be54a6f9f6c0dbdc5fbeaca : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6fb765cf2e614391a4853602d32b65d9 : SUCCESS in 1h 16m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2fdcdd5c6dae44508a4b8e47da72599b : SUCCESS in 1h 24m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91441fc8d55b483588fecd998641f9ff : SUCCESS in 30m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0bde54e18b114624a31da39e45596cdb : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f92e3605d4442c084073ced0a2582b2 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f7946a5ea2db2e7eb66b33e5ce9123e138b9a3bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:52:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"681a1401c052b5ebe4fdf753adb58633cbdf49a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:37:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/610a7b1639e94afa924bff0be0f9a2be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6599469b694693be9202ffd0015b61 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048c1d95480f42a5aeb9149a37d3b404 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc3f2fb6a80a4fc3b93c0c89194696f2 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0dc9c770737417aa9d40845f0c1a4ba : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d09586e862e9457d85fc5561082e4304 : SUCCESS in 1h 42m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/364cd283d7f34fa28be2fcdc9922b61f : SUCCESS in 1h 09m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/42de597538ba43ac86a095b1951ab6c7 : SUCCESS in 1h 37m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9dc812ebb867429c91cf64ac5fff72cd : SUCCESS in 1h 25m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/781808495e914de393b99a09403f4017 : SUCCESS in 1h 10m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5400e7ab10c4c759fae55f0745514bd : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"075f352f6b57d5bc624b25628c8bf3cbd25ebfcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:37:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fed713201ab6070b51acd961b7675bbb3b6e76ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:38:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b21b0b119f2472585658373cddd50bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b18d4c7d244d9bbbcf218dd7034613 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23b1fbc3b37f2b53061500b2a5034cf5d074abad","revisions":{"23b1fbc3b37f2b53061500b2a5034cf5d074abad":{"kind":"REWORK","_number":1,"created":"2024-02-14 18:33:48.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/39/909039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/39/909039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/39/909039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/39/909039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/39/909039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/39/909039/1"}}},"commit":{"parents":[{"commit":"35455d101bb09ca71620216f9b573baa5666a961","subject":"Merge \"Replace CRLF by LF\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/35455d101bb09ca71620216f9b573baa5666a961"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-02-14 17:34:04.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-02-14 18:33:23.000000000","tz":0},"subject":"rpc: fix some typos","message":"rpc: fix some typos\n\nChange-Id: I3669e11839a99e7d1f60b5be16f1097295196628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23b1fbc3b37f2b53061500b2a5034cf5d074abad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23b1fbc3b37f2b53061500b2a5034cf5d074abad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
