)]}'
{"id":"openstack%2Fneutron-lib~909042","triplet_id":"openstack%2Fneutron-lib~master~Iabcf6e43fca40e7dbdd83345b2cd6d32cb25d3d8","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-02-21 01:12:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabcf6e43fca40e7dbdd83345b2cd6d32cb25d3d8","subject":"rpc: initialize notifier when get_notifier is called","status":"MERGED","created":"2024-02-14 18:33:48.000000000","updated":"2024-02-21 01:13:29.000000000","submitted":"2024-02-21 01:12:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909042","meta_rev_id":"a37f062365ad0e6f28e2a2c342a03f04aec9b858","_number":909042,"virtual_id_number":909042,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 01:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-16 07:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-20 22:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-20 22:55:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 20:25:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 07:13:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-20 22:55:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"72714d92ed05baf96397d8c6bbb772d127e56476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-14 18:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913a3b13a38cbf7dcb686cedb84cf1ff9549eadf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 20:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b427f64252e471dae64b3bf5cf13d63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e800e489e749ac83f219833aa06184 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c102c2399ea74ad59009ca3cd5084bda : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f046247ddbf41ce9372e027772ae19a : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78a006e9d6f64130a8b4227554c70181 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4933603139b44ebaa203394eed4efdae : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5965a20143b4d8d8e3a6b743b59d83f : SUCCESS in 1h 36m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a0ee26d1661248a6ae27affab1b472cf : SUCCESS in 1h 19m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6a5fb125da443778d60b8b95ce32406 : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6d205eacf0614905a84a26bf6064c96c : SUCCESS in 1h 30m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f7047cce892c4ea38ae937b8347e329d : SUCCESS in 1h 48m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/276c5f38a5004c9d93eaac09ea6695ca : SUCCESS in 37m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4b6902f2c06747518aee0e70849330df : SUCCESS in 6m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de448f034d824932939b797b304f8a8f : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e17a544d04e12590d8e9b18805a2e5cddf86a44d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-16 07:13:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8bce49ba3928e91a5f53a5526acfc2716be34c39","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-20 22:55:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f28cae9970baf4477d472d6c4e45e4b5e1af75f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:55:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4e9507652fdb71efa2ac8817cb86fa02b4f5d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 01:12:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72ad6cefcf8408d948a1ea951e9ba95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d42832c3f85c4a24bd662fbb72aff7b0 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d9bae9bfd54ecfa8ef55a2bbacd4fe : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/549b9cfec3b04142a29536c56b39dadf : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f26a5cb12354304a595fca59ea4f1cd : SUCCESS in 5m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/574a30204f304573956ad0e423b2f008 : SUCCESS in 1h 46m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/abee26aabb0d4e54b3fab08fa6d875a8 : SUCCESS in 1h 42m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e52eb7e77880418abf6ea5cbb62e2f26 : SUCCESS in 1h 43m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d77c8083925647fe9fcf6de126aa963b : SUCCESS in 2h 13m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/da56de5962d64d3c89df5e914d6f5e07 : SUCCESS in 1h 31m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcb11e981859492fa27f7be5e9c7fb5e : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":1},{"id":"78a53a1e2865102d029d0e38fb39fe4199f5f4ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 01:12:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a37f062365ad0e6f28e2a2c342a03f04aec9b858","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 01:13:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47eacf6a55504c478e35c2e066a99dde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bc5b809b7647b49053aa75abe89285 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"631e3f995c6efdf6596bf5dad406de4143c37b4a","revisions":{"631e3f995c6efdf6596bf5dad406de4143c37b4a":{"kind":"REWORK","_number":1,"created":"2024-02-14 18:33:48.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/42/909042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/42/909042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/42/909042/1"}}},"commit":{"parents":[{"commit":"1fedbae8336aa077cdbc1ba6e2819f5da04cd39c","subject":"rpc: don\u0027t explicitly pass access_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1fedbae8336aa077cdbc1ba6e2819f5da04cd39c"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-02-14 18:07:25.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-02-14 18:33:23.000000000","tz":0},"subject":"rpc: initialize notifier when get_notifier is called","message":"rpc: initialize notifier when get_notifier is called\n\nThis is to mimic other get_ (server, client) entry points.\n\nChange-Id: Iabcf6e43fca40e7dbdd83345b2cd6d32cb25d3d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/631e3f995c6efdf6596bf5dad406de4143c37b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/631e3f995c6efdf6596bf5dad406de4143c37b4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
