)]}'
{"id":"openstack%2Fneutron-lib~909366","triplet_id":"openstack%2Fneutron-lib~master~I0b2f2225e29537c9fd2de53b0945a451b9bcdde3","project":"openstack/neutron-lib","branch":"master","topic":"bug/2052937","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-20 01:05:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b2f2225e29537c9fd2de53b0945a451b9bcdde3","subject":"[S-RBAC] Add note about port:binding:profile field and SERVICE role","status":"MERGED","created":"2024-02-19 08:10:44.000000000","updated":"2024-02-20 01:06:06.000000000","submitted":"2024-02-20 01:05:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909366-bug/2052937","meta_rev_id":"b15c879480546aea440ccf67f63ef415ad1181c5","_number":909366,"virtual_id_number":909366,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-20 01:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-19 08:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-19 08:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-02-19 14:33:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-19 08:17:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-19 08:25:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-02-19 08:32:52.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-02-19 10:01:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-19 21:33:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"ba84f401cb05ca8ce98f9be172b725cb5f7a3fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-19 08:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63b8bdc80740f4484802fd2ec792d12f1001e759","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-19 08:17:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c6d89991fd959821e64af2dedeec5d9b670e6d2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-02-19 08:32:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df30416f0bedeed45acae7a5c535847bded1e362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 10:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa5d3c2cab624701b011b7e24d1cf77a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23798aa30def42cfbcb21089b2d34a53 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ceaa1d5a4704a3894b49d99a8d98488 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0983b0ca8aed4c8f8ba1c0d115fa4259 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ad1d8c58aa6a450e90837a818c92d17b : SUCCESS in 1h 05m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/70ce46723bda4beca5f6433504544b61 : SUCCESS in 1h 18m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/989be0064ab341b7aec0eba6335d2c1e : SUCCESS in 1h 09m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9ef63a4b145642508ab7db16b0b6e17b : SUCCESS in 1h 44m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/20fb094c1c4740a3a7fa79fe3bcacf03 : SUCCESS in 31m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/308b81fd525844a38261bb14ba70e644 : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bb85fe76e44bfbe367c283f884f50b495f79d2d1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-19 14:33:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"789f925d6638c14959ec9cd957cec9f8283fdb5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:33:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e7aaf58a5a843da312ba35e3ff29b6442b71f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:41:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1abfca911af242848cb7a9dab777c26c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fa3fd91b1a144a9f8d42f6d3ebfe4b89 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1ca12900e549509f7a8e6e37c0aa79 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80002c37160649f2ab21fb0c79d56b15 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/350bbac788434cbf805751d20125e1ff : SUCCESS in 1h 56m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a5a0b7e89cb499681c540584396efe1 : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f228836fe6734eb2a30b13adb4f541d3 : SUCCESS in 2h 06m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b652413f58eb4ec7949473ac4ed82b1d : SUCCESS in 1h 21m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2b53405175d476eb9bf9460e0a2a693 : FAILURE in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"701815f9ce53d6b9ab7a9d890ae97ad7cefacf98","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-19 21:33:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc9d29dc089f37d62edf95df7d53c1363a279045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 23:04:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10893c2b1007451d8606f9bae8bb83bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d35de5f85f90410884e88371d39dce27 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90804368f78d4616b5e90471b40bf2b2 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8406e50f7a9947969a9f3817a94e86ba : SUCCESS in 4m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/934c2417c8b1448fb52ca274418886dd : SUCCESS in 1h 23m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91e41f80f8a643399f1a52e4563e4767 : SUCCESS in 1h 18m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/eb9a253ea634455ab2e879d059bc9c8e : SUCCESS in 1h 13m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/725fd3da086947249da2685fcc823cdc : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6aed5ff751354aafbf375672c454928e : SUCCESS in 32m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73c3f052b56f48c995877f4591335a4e : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":1},{"id":"68d7a1c65a0d843c196318fc8d9a983621ab9d11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 23:04:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42e8319f908cc65719a16b4d96f5c54af0d16896","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:05:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bea4f41eedab4e49955dde175b9453f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f566efd3b0184fb8afd928b2bc8fc957 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb3cce3dc1d4d3e805ad2a08b7a6564 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45613d9449248e19373f05951863298 : SUCCESS in 3m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8979e6c53a1b42d4b7f3404731a13ffc : SUCCESS in 1h 22m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43e13763b0e546cba912865ee0e8ccc3 : SUCCESS in 1h 34m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c91e49ab43f47c2a09d7ee73a104489 : SUCCESS in 1h 55m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b75843bda5f94d06a4f8941e3f1d3f43 : SUCCESS in 1h 27m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/917124d4c5e840e3962aab0d0eb1bed4 : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":1},{"id":"eb3e0c6f26beac292c3332e2509b01cb83c6b127","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:05:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b15c879480546aea440ccf67f63ef415ad1181c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:06:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a52858417945849b0faab473467d9a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d25d37e282584351945e4b416ddb4a06 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b5fd609ee244749a09c03f7b8b262e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3aec8fdfeeb469419c9a6760355e203a1bc8a790","revisions":{"3aec8fdfeeb469419c9a6760355e203a1bc8a790":{"kind":"REWORK","_number":1,"created":"2024-02-19 08:10:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/909366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/909366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/909366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/909366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/909366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/909366/1"}}},"commit":{"parents":[{"commit":"1fedbae8336aa077cdbc1ba6e2819f5da04cd39c","subject":"rpc: don\u0027t explicitly pass access_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1fedbae8336aa077cdbc1ba6e2819f5da04cd39c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-19 08:08:54.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-19 08:10:40.000000000","tz":60},"subject":"[S-RBAC] Add note about port:binding:profile field and SERVICE role","message":"[S-RBAC] Add note about port:binding:profile field and SERVICE role\n\nWith new default API policies binding:profile attribute of the port can\nbe only set or updated by the SERVICE user. This patch adds small note\nabout this to the Neutron API-REF document.\n\nRelated-Bug: #2052937\nChange-Id: I0b2f2225e29537c9fd2de53b0945a451b9bcdde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3aec8fdfeeb469419c9a6760355e203a1bc8a790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3aec8fdfeeb469419c9a6760355e203a1bc8a790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
