)]}'
{"id":"openstack%2Fneutron-lib~909942","triplet_id":"openstack%2Fneutron-lib~master~I2b0cfe3f726b242f602dfc98b738030e7cdcf556","project":"openstack/neutron-lib","branch":"master","topic":"bug/2054324","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-03-06 18:44:33.000000000","reason":"Change was abandoned"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-03-06 18:44:33.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-03-05 16:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I2b0cfe3f726b242f602dfc98b738030e7cdcf556","subject":"Fix iptables mapping of \u0027ipip\u0027 protocol","status":"ABANDONED","created":"2024-02-22 23:37:39.000000000","updated":"2024-03-06 18:44:33.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76db8748cd7465f7ffa5eaa16402f4a36fff7db7","_number":909942,"virtual_id_number":909942,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-27 22:05:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-29 07:37:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-03-01 05:43:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":-1,"date":"2024-03-05 16:12:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 01:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 14:24:05.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-02-29 07:37:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-29 19:09:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-03-01 05:43:08.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"80c12dfed985945dc712cb9625abacb6e032584a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-22 23:37:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4af77c65ace5f766859a09eb2e7e398af6ed3df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 01:48:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a59b054c2648389c5fe5c34dec039b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6257cea49a470bbad516ced08b690f : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecc930edeee843199f5da26a33a1c7ba : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae26f5c413c4eb2bbb1329c82b0da57 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eeed4d8ffa64ae2b091b88ffae75ece : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b573360d2824d3bb9c50a9415a341fb : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/382028fe2804489298b0be2eb40a1d1b : SUCCESS in 1h 06m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e2ed8745639c4e7d98e93d6f022e4068 : SUCCESS in 1h 41m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9a0b8c588f94e60a0f01436f72b3aa9 : SUCCESS in 2h 08m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f8799b24aa6b4fcdaa3ae7982361e189 : SUCCESS in 1h 40m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aee1446fcf2d4a35940cc9ab1685fb65 : SUCCESS in 1h 55m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ab1b45f0bb345c0a6a676445384833a : SUCCESS in 42m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0bb14b1b78cd41d39486cf4b65b3c990 : SUCCESS in 6m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2004b58d09c4691a1466af37e64e08a : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":1},{"id":"df99521f3949fd594ff3ff80f9fd35210003904c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-27 20:14:30.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"389756b16f2674bc23f2d7f7db5bed563b61d813","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-27 20:15:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f5b51b9a5eb287da4f7e809208e462fae7f8d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:05:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a637b6837617428880375428f3c7ad27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4d2fc611614397839b3654cd652ad3 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/178dd1e795834223b70d5d606ad1508d : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66965d5b3fac4115adb36b16ca8dbe5d : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b1941ff9cf54c97b84f40d514a9f3e6 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1347b7c477af4a66a9c48250a47a2fe9 : SUCCESS in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a692ea69110144a9ad265e522b6267f9 : SUCCESS in 1h 40m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bfd2bf3a8cd24644857510653294646d : SUCCESS in 1h 29m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33198b2b1b5e4bd3ad64a90d52533c98 : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/80a2c562b39843e9aec73e21872304d5 : SUCCESS in 1h 29m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/08c2b5c3328748d9b673db72f09adc03 : SUCCESS in 1h 17m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebc2c6d9a15d49c1a33572c9669465cd : SUCCESS in 32m 53s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3140c16def964278b4ec8de8e9ada134 : SUCCESS in 5m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a574717924374edab5c8d3fb2e7011cc : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9cee70a05d4d19f8fdefabbc2c0786d1c93c095a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-29 07:37:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8dfa32639a9b62d9d24c6522190cfdb256dd63f1","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-03-01 05:43:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b10f2415d9d3a93b50f80c91c1dec8b294173993","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-05 16:12:23.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a395ffe3db7fbb8414743225f72cd634231adddf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-06 18:44:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76db8748cd7465f7ffa5eaa16402f4a36fff7db7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-06 18:44:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b3e6a780c199ba5890ede243185dab298beff88","revisions":{"82c788a59a07df787d2f977660073b0ef0990197":{"kind":"REWORK","_number":1,"created":"2024-02-22 23:37:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/909942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/42/909942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/1"}}},"commit":{"parents":[{"commit":"8ce8b7214e68bf43ef13086942f1222a5be39b0a","subject":"Merge \"Remove constants for LBaaS(v2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ce8b7214e68bf43ef13086942f1222a5be39b0a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-02-22 23:25:04.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-02-22 23:25:04.000000000","tz":-300},"subject":"Fix iptables mapping of \u0027ipip\u0027 protocol","message":"Fix iptables mapping of \u0027ipip\u0027 protocol\n\nThe neutron API accepts \u0027ipip\u0027 as a security group protocol,\nwhich maps to IPPROTO_IPIP (4) in in.h.\n\niptables-save uses \u0027ipencap\u0027 when \u0027ipip\u0027 is given as a\nprotocol argument, so enforces an incorrect rule.\n\nMap \u0027ipip\u0027 to use the string \u0027ipencap\u0027 so the\nIptablesFirewallDriver class in neutron works correctly.\n\nChange-Id: I2b0cfe3f726b242f602dfc98b738030e7cdcf556\nPartial-bug: #2054324\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/82c788a59a07df787d2f977660073b0ef0990197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/82c788a59a07df787d2f977660073b0ef0990197"}]},"branch":"refs/heads/master"},"5b3e6a780c199ba5890ede243185dab298beff88":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-27 20:14:30.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/909942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/42/909942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/42/909942/2"}}},"commit":{"parents":[{"commit":"8ce8b7214e68bf43ef13086942f1222a5be39b0a","subject":"Merge \"Remove constants for LBaaS(v2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ce8b7214e68bf43ef13086942f1222a5be39b0a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-02-22 23:25:04.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-02-27 20:11:54.000000000","tz":-300},"subject":"Fix iptables mapping of \u0027ipip\u0027 protocol","message":"Fix iptables mapping of \u0027ipip\u0027 protocol\n\nMap \u0027ipip\u0027 to use the string \u0027ipencap\u0027 so the\nIptablesFirewallDriver class in neutron works correctly.\n\nLong description below.\n\nThis is one of those confusing edge cases and I think\nLinux is conspiring against us. Let me explain.\n\n1) neutron-lib does correctly define the protocol name \u0027ipip\u0027 as 4.\n\n2) The linux kernel uses the same in in.h:\n\n IPPROTO_IPIP \u003d 4\n IPPROTO_BEETPH \u003d 94 (?)\n\n3) iptables maps \u0027ipip\u0027 to 94 and \u0027ipencap\u0027 to 4.\n\n # for num in {0..255}; do iptables -A INPUT -p $num; done\n # iptables-save | grep -E \u0027ipip|ipencap\u0027\n -A INPUT -p ipencap\n -A INPUT -p ipip\n\n4) /etc/protocols does the same as iptables:\n\n grep -E \u0027ipencap|ipip\u0027 /etc/protocols\n ipencap 4 IP-ENCAP # IP encapsulated in IP (officially ``IP\u0027\u0027)\n ipip 94 IPIP # IP-within-IP Encapsulation Protocol\n\n5) getprotoby{name|number} does what /etc/protocols does:\n\n $ getprotobyname ipip\n struct protoent: (0x7fbbbcca9c60)\n   p_name ipip\n   p_aliases IPIP\n   p_proto 94\n\n $ getprotobynumber 4\n struct protoent: (0x7fc51ad86be0)\n   p_name ipencap\n   p_aliases IP-ENCAP\n   p_proto 4\n\nNeutron actually builds a mapping based on the getprotoby*\ncalls, so in the iptables case it winds-up doing the wrong\nthing.\n\nPartial-bug: #2054324\nChange-Id: I2b0cfe3f726b242f602dfc98b738030e7cdcf556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5b3e6a780c199ba5890ede243185dab298beff88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5b3e6a780c199ba5890ede243185dab298beff88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
