)]}'
{"id":"openstack%2Fneutron-lib~911392","triplet_id":"openstack%2Fneutron-lib~master~Ib66743037166c3390b05f4d43e39eac397ef156b","project":"openstack/neutron-lib","branch":"master","topic":"deprec","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-10-04 10:18:17.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-10-04 10:18:17.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-28 18:04:09.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib66743037166c3390b05f4d43e39eac397ef156b","subject":"Deprecate get_notifier(service\u003dNone)","status":"MERGED","created":"2024-03-06 03:12:21.000000000","updated":"2024-10-04 10:19:20.000000000","submitted":"2024-10-04 10:18:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911392-deprec","meta_rev_id":"80f35311e8ad203e8ed0dc245c8edef3eefbc508","_number":911392,"virtual_id_number":911392,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-04 10:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-10-04 08:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-26 08:00:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-08-28 18:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-10-04 08:05:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 05:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 16:17:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-07-03 07:03:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-08-26 08:00:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-08-28 18:04:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-04 08:05:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"73126bf552b8e32e618089727354ee2572b3bc0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-03-06 03:12:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb8009b3e39a85207fee9ffab9ced5c8a552ff8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 05:02:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7320a5783bb34521bf2cdb656fd4bd05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6648ae15674e7587b50cbbd71a418d : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af8a4a58c7474599a2b52131938fd07a : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1795184078c4426da9c6e46f6586c77e : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab750dce0024d94b3be0d9d6573c46d : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40dbf6f5b2b341639fff6a6689f4551f : SUCCESS in 4m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1eeb70f5e054c9b98603e08df1b4e60 : SUCCESS in 1h 43m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9972462708814e0ea924a6dace1be5e4 : SUCCESS in 1h 28m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66dbcd42c3f74d5bb6b8b73083af85f8 : SUCCESS in 1h 37m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d498f1602bfc4a1eb4f3177f76b53ce4 : SUCCESS in 1h 28m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d3cadd0828314bd19858cbd1c57602a7 : SUCCESS in 1h 48m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4126cf743c634ed3aa10ff864a6d80e7 : SUCCESS in 32m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b55c836ec08444019653e1bc9402e6c2 : SUCCESS in 6m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d533ebed02994458870558a71f2eede5 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ca2429df2c4557b0a4cb7f333436f157b195abde","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-07 16:17:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5ebd61968526e0f79a0239416b6e7a8ada7a7f95","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-03 07:03:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0489ccc8431c9538287d442707ed055f1c5555f9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-03 16:09:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"194e79f3deac641e4c623c5246b627ecf5144756","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 20:51:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54feb49218d770aca86cb007feece921945ce099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 20:51:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b4c431d41b794849a24180f1f1dbfb09b89ce69a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 23:43:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac78613dd9524d3d93043723f1768b84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6273870e06cd4d3f8122de6af1bfabc6 : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c4c3c0729214a608afc409c400a1cf5 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc6a72edd674e1bb72e7d1c655c2886 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab9739f2148843ca98a87766de71d5a4 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e68514eea71476286280d7d25c79bac : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4086c5a056264a9e99bc33d48b21485a : SUCCESS in 6m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa290c436be243bb86e670b86834d0ba : SUCCESS in 1h 39m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ef43904357418d853f3a312eb8e4f7 : SUCCESS in 4m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fb87c026a76e4f91bbf2d09dca8d348d : SUCCESS in 2h 00m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a17c1fb010c34c11aae4a04fdecdcb29 : SUCCESS in 2h 13m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f462472fbaaf4b049882c65e9e084b39 : SUCCESS in 2h 05m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94cfa3f0b5ea43be996557ed0305f211 : SUCCESS in 2h 45m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3ee0bd8037b49a4a1d921685db40904 : SUCCESS in 33m 24s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/492cba59f8314070940f96040c4725fd : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1f8ad83428bf453c9d46e00268f21d20 : SUCCESS in 6m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b0bc5bfa04340319d99312470e0d089 : FAILURE in 47m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ee8b7cb7eac2e3ba9578e672d31b5596a5591039","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-26 08:00:30.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"92e8196593c7af5392edd07d8dde96705cb80d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 10:03:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bfff34485ee405c85be64ea8ac17547\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c80085082d1463cad85aa1162643a6a : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80f711c4acc84410a59cac86216c936c : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516ebc7523b6409eb53de3fa1b35215e : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d45e0a14cfa4b80a391f5c4b874a5b8 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8971224e972482f896e529a03a50bd9 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a31bc786f764ba4978cd547ffcafdaf : SUCCESS in 4m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/495dd925ff274b67b1dde5ab06fbc82f : SUCCESS in 1h 31m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33fdf8db9d08472b9b1080d60304f99c : SUCCESS in 4m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d492b5f8dd14d5e8e34138350a70e52 : SUCCESS in 1h 25m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78b3c99d35064621820d0ceefd4a0ecc : SUCCESS in 1h 32m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f28a797a093f47d4bf27bfe9a0365988 : SUCCESS in 1h 19m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b3a7a980a846413e83557a5b8213400e : SUCCESS in 2h 01m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/344b34a2f7b8451daa1aa925ab34dc24 : SUCCESS in 33m 24s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/9f2270e6de24420ba37de49184ef73cf : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/37569f97b8d34a8fb842b0e9be46f1cb : SUCCESS in 5m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a095dfbfda0c4375a7ac9824c3d0b220 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":2},{"id":"285ea0d777dd063deba024bd3438ec438a3a51e1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 18:04:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e226e82697cc864ee38767363a9a72aa9da6f4e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-30 14:53:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c1fb71c2846a27d397eab688e171915cf3c9cd8f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-04 08:05:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a35d80b2d5f9c640e6048b69936ec7410a1efe4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 08:06:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b03a8aaad71f3a009ab4b0f2307d3e5038173a03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:18:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dd9a1da072a4d0b8f818bcde2320e46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29d23669f094d07a11b07d4aacce0a7 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dc3521e8427404d84c0a502b1d57cae : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d9b1a122cff406098eb8fdbafdd070d : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/764836f8b4634d70bc6d3ef6fe15d339 : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fe1be933597420fb8463912a945d3fc : SUCCESS in 1h 17m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09886c0dd80343158ba9de2a4e193d66 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7526020748d7478fbad9cd1a71bd288e : SUCCESS in 2h 02m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ad274f6436649a09263b455a4da235c : SUCCESS in 1h 21m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1dcb1bb6d8f846d4b8d8443ddbc9df09 : SUCCESS in 2h 11m 48s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a5bd8a617e654f4ca409d1813ac4c0a4 : SUCCESS in 1h 25m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21bae6d2865747b585ed4ecdb8b8128b : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":2},{"id":"81c307fa5a5fb0d1fa8399bba821be9251173c3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:18:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"80f35311e8ad203e8ed0dc245c8edef3eefbc508","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:19:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b04d2e696e6430f8b475e40f8fbf969\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/398c8da3f7df432bacad6413e7e7c9bb : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89c497d535844f048a5b7a9b8768e1df : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d37cbfebdf62d00a27bb11cf90a2137e9f0fd1e","revisions":{"a2fb237061a61a89d8840c44c8c2e82946c6f5a6":{"kind":"REWORK","_number":1,"created":"2024-03-06 03:12:21.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/92/911392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/92/911392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/1"}}},"commit":{"parents":[{"commit":"3a9e39b9ef4d73cc60b534411bc5255e7d280bef","subject":"[netaddr\u003e\u003d1.0.0] Do not use netaddr.core.ZEROFILL flag with IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a9e39b9ef4d73cc60b534411bc5255e7d280bef"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-03-06 03:10:28.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-03-06 03:11:33.000000000","tz":-300},"subject":"Depreate get_notifier(service\u003dNone)","message":"Depreate get_notifier(service\u003dNone)\n\nAll known consumers already pass a name and don\u0027t rely on the behavior\nwhere \"None\" (string) is used as a service part for the notifier topic.\n\nThis change is a follow-up from discussion in [1].\n\n[1]: https://review.opendev.org/c/openstack/neutron-lib/+/909040\n\nChange-Id: Ib66743037166c3390b05f4d43e39eac397ef156b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a2fb237061a61a89d8840c44c8c2e82946c6f5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a2fb237061a61a89d8840c44c8c2e82946c6f5a6"}]},"branch":"refs/heads/master"},"3d37cbfebdf62d00a27bb11cf90a2137e9f0fd1e":{"kind":"REWORK","_number":2,"created":"2024-08-22 20:51:49.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/92/911392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/92/911392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/92/911392/2"}}},"commit":{"parents":[{"commit":"3a9e39b9ef4d73cc60b534411bc5255e7d280bef","subject":"[netaddr\u003e\u003d1.0.0] Do not use netaddr.core.ZEROFILL flag with IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a9e39b9ef4d73cc60b534411bc5255e7d280bef"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-03-06 03:10:28.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-08-22 20:51:48.000000000","tz":-240},"subject":"Deprecate get_notifier(service\u003dNone)","message":"Deprecate get_notifier(service\u003dNone)\n\nAll known consumers already pass a name and don\u0027t rely on the behavior\nwhere \"None\" (string) is used as a service part for the notifier topic.\n\nThis change is a follow-up from discussion in [1].\n\n[1]: https://review.opendev.org/c/openstack/neutron-lib/+/909040\n\nChange-Id: Ib66743037166c3390b05f4d43e39eac397ef156b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d37cbfebdf62d00a27bb11cf90a2137e9f0fd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d37cbfebdf62d00a27bb11cf90a2137e9f0fd1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
