)]}'
{"id":"openstack%2Fneutron-lib~911501","triplet_id":"openstack%2Fneutron-lib~unmaintained%2Fvictoria~Ia31bf1418d499ec8df5df4c8b7e126c03a64d5ce","project":"openstack/neutron-lib","branch":"unmaintained/victoria","topic":"create-victoria","hashtags":[],"change_id":"Ia31bf1418d499ec8df5df4c8b7e126c03a64d5ce","subject":"Update .gitreview for unmaintained/victoria","status":"MERGED","created":"2024-03-06 11:58:22.000000000","updated":"2024-03-06 16:44:02.000000000","submitted":"2024-03-06 16:43:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911501-create-victoria","meta_rev_id":"245906bcf27c1e3f64cc55a819583b41675eba52","_number":911501,"virtual_id_number":911501,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 16:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-03-06 15:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-06 15:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 12:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 15:47:00.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"93a4df563e727e36047aa233b219cab7047282c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-06 11:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0047e6d6b52da47ec189c60a6e90aafd4ae02432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:56:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd70585293b40c2bb28c743e7cadcb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e53512e57e14a13a883bfeb9136abef : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1dcb25a63274c9188fb8e1155a44a01 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d75f140c5354e28b9bc1fa331f4c4be : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56207112d59448cd82a0c5b9fabfdf2f : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eac64db20faf4604aafd16906b6137e6 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/91daea29de8840c28e0e83b78f67b8e1 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ea8a46513cbb6ba10b9d39f4868fba1f83868f2a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 15:47:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b35433ef141d35e8bd6ccb0ecfb32a8654a1c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:47:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd5086f14d0e0c81a209c82b6dd6ed184d9d499d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 16:43:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abb38d1c492b40229fb9a540ad46e21e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f95290058004c3295ab66cbe5aeaa32 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bcef30b6d54c80b99a712a61a1b1f3 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39fae36cf9934e1da131483c4808df0b : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afdca125dfdf4900be6a4b8e5c6e1b6d : SUCCESS in 4m 36s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/9e045f0a3d0e402fb3d244d149dca82e : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1ca1fde9dee0ed48ace92c4150eeedc30b4f2cfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 16:43:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"245906bcf27c1e3f64cc55a819583b41675eba52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 16:44:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908a1452a39b4e2b809b03a7a4f1c0ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d107f777ad50407a8a2e275a2ec6a417 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be45cada63f8dd9c79837da807629b52a9bf8068","revisions":{"be45cada63f8dd9c79837da807629b52a9bf8068":{"kind":"REWORK","_number":1,"created":"2024-03-06 11:58:22.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/01/911501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/01/911501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/911501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/911501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/911501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/01/911501/1"}}},"commit":{"parents":[{"commit":"0b9d8e399f845999668a9d7ceaa3a638fd1c399f","subject":"Enforce policy for qos_policy_id attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0b9d8e399f845999668a9d7ceaa3a638fd1c399f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-06 11:58:19.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-06 11:58:19.000000000","tz":0},"subject":"Update .gitreview for unmaintained/victoria","message":"Update .gitreview for unmaintained/victoria\n\nChange-Id: Ia31bf1418d499ec8df5df4c8b7e126c03a64d5ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be45cada63f8dd9c79837da807629b52a9bf8068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be45cada63f8dd9c79837da807629b52a9bf8068"}]},"branch":"refs/heads/unmaintained/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
