)]}'
{"id":"openstack%2Fneutron-lib~914826","triplet_id":"openstack%2Fneutron-lib~master~I895cc9aaee2546b95e93150c134f350a50346447","project":"openstack/neutron-lib","branch":"master","topic":"fix-filter-spec-link","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-09 09:43:04.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-04-15 09:16:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I895cc9aaee2546b95e93150c134f350a50346447","subject":"Fix incorrect links for the filtering spec","status":"MERGED","created":"2024-04-01 18:43:30.000000000","updated":"2024-04-15 09:17:19.000000000","submitted":"2024-04-15 09:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914826-fix-filter-spec-link","meta_rev_id":"bd6a38875b5aed0d3d61541a7157518919db3db5","_number":914826,"virtual_id_number":914826,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-15 09:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-09 09:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-04-15 07:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-15 07:34:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-01 20:22:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-02 05:14:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-04-04 14:14:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-15 07:34:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"46b5132cd6776a7bd4f992c498c6c699591b64cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-01 18:43:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec69778b9e0119b019603fdf46aa9a86f173faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 20:22:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5536d006a1f47faae0cf738fafe37dc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/051ae76c676d40dbb56500fef0392b56 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f0a4eff5364fedaf595345047d97dd : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45bc10ce670d4cc6be4394de65faf7a1 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e1097b8b50041c0934b23700e4afdbe : SUCCESS in 1h 07m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d95e318e46a14cb2b868f1f261b5c739 : SUCCESS in 1h 37m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2071008e6967437196fa28cfe549882c : SUCCESS in 1h 18m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1e0a16e2ff77499aaf217daa0485ffd4 : SUCCESS in 1h 24m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cdf2121015c8413fa9c238ca51703299 : SUCCESS in 31m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7de0fe7f96cb42328576651484dff23a : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7db723e74f1c780b277021727e7792ffbc438107","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-02 05:14:30.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"49e5b0d8307c6be2a014db6cdf681561c6b40710","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-03 00:29:20.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b01767826f67c8343138eff91a0185496cb782d9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-03 00:29:24.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"78dfa9338aab572f0815c5fddeb0721f34b7e9eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-03 00:34:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98a84ff753a6e3bd82d813d139a2976600315c87","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-03 01:19:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7e8d59f70eb32a72da308df5c16e19ff3f84ecb2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-03 01:19:56.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"a97d1ad2e99bcbbba93b5bafdaae6602189bcee1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-03 01:20:06.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"06ae15fffd6f02f9ec605880adaaa0c41fccc368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 03:19:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c731ddd53443450eaf19f557ccc8410f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/529dd1c36365485c875bad48bd6843c9 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f9ce0722f242e18757afcf5c6557d9 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a585c60c4eea4df39ad8176859dcf540 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/520394b008e54b37ae1c1a35b232ce40 : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7766d0e92e304afda13553bcdd3f276a : SUCCESS in 1h 42m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/cba62ef05e8c4455b5661a9fcfa6d6d3 : SUCCESS in 1h 33m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/702d571559af4647aaee7e61c809b4e5 : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90548b9b22664483be0a3c545045d546 : SUCCESS in 33m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60891dca7c624403859e587405e500e0 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":2},{"id":"75ff39118bbc67e026e510bed72002e21b110317","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-04 14:14:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a12f84375f23980dab18e262e57243733df3fba6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-04 23:00:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a07a0a962707668624acb7b218af7d2d124d099a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-04 23:00:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"892617ab2c5b60b429bd547f08a4c23c4596afdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 01:06:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb54d9651c3d4903834a2450d0461c6e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/379e768adff045d2bb4834cd0f526da1 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41803c5c68a746488f34438229a7b5b7 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28b7d3249744dd985e5a0f89cc1700f : SUCCESS in 4m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f9adfbfb4e8f406d9816b268434ed50d : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0072988d0464459e8c2302d50b6a9929 : SUCCESS in 1h 55m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/991b172adbee4867b795f12ba3314a25 : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc7d5c63a2fb4615b6d8956e3bc19ccc : SUCCESS in 1h 42m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c8c9e21a32b4f518bb2adf7195b3321 : SUCCESS in 36m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f9e0ae8f154400ebd4bebe187630c71 : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5320cc87fd26f4af6bf963410899ffd14b6252d8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-09 09:43:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96683317b0f803db427fabb3555e7e462b289f46","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-15 07:34:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54433f29507d61ad1e4cc1fa4d9bd9e31ab74d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 07:34:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d529bfb2a78f88a8634c52a553e191b8ab0a5c40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:16:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c4f641d90245cb9e1a6a5058f51418\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/32fad2c8826c422485098ac3fee897fc : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55dda9c48d434844bea42990856c08a5 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82cb2eb7367428eb081b810bea61cc4 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f41dc2d549e48f68c51dec58fefb050 : SUCCESS in 1h 13m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e5e26620bf1410e921f11f7b86111bc : SUCCESS in 1h 40m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c68706159a194ebe91688f034b7720cc : SUCCESS in 1h 32m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/bb7ad6fc3c544d39b8133e80ee291778 : SUCCESS in 1h 14m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/44258643de664cf686021d8040185174 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":3},{"id":"80ae860838ea4760f6bbef857183f8e346cf0f07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:16:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bd6a38875b5aed0d3d61541a7157518919db3db5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:17:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6154fd17956f4dfb8c2ed33c2017c71a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/016a527f0571424d81b4861ca785d5a7 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c29a6967e147a58b3d03e64d080cd3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c28b3ab8fe5674887b643ac76dec2ed9837313d3","revisions":{"b213b4297729480b5d90b2688014b9b0839f2068":{"kind":"REWORK","_number":1,"created":"2024-04-01 18:43:30.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/914826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/914826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/1"}}},"commit":{"parents":[{"commit":"b4cc4bb090fb54de837546972f57bad5644daf67","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b4cc4bb090fb54de837546972f57bad5644daf67"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-01 15:17:46.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-01 17:12:43.000000000","tz":-240},"subject":"Fix incorrect links for the filtering spec","message":"Fix incorrect links for the filtering spec\n\nThe spec networking_general_api_information.html was moved\nto api-ref in [0] but some of the links pointing to it\nwere not updated. This change makes all the refs consistent.\n\nAlso moved sync/async and bulk create sections lower to\ngroup all the filtering, pagination and sorting sections\ntogether since they are inter-related.\n\n[0] https://review.opendev.org/c/openstack/neutron-lib/+/362877\n\nChange-Id: I895cc9aaee2546b95e93150c134f350a50346447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b213b4297729480b5d90b2688014b9b0839f2068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b213b4297729480b5d90b2688014b9b0839f2068"}]},"branch":"refs/heads/master"},"a44f91352a9ef7b5d414f94fdcbee6ef0eac621c":{"kind":"REWORK","_number":2,"created":"2024-04-03 01:19:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/914826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/914826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/2"}}},"commit":{"parents":[{"commit":"b4cc4bb090fb54de837546972f57bad5644daf67","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b4cc4bb090fb54de837546972f57bad5644daf67"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-01 15:17:46.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-03 01:19:00.000000000","tz":-240},"subject":"Fix incorrect links for the filtering spec","message":"Fix incorrect links for the filtering spec\n\nThe spec networking_general_api_information.html was moved\nto api-ref in [0] but some of the links pointing to it\nwere not updated. This change makes all the refs consistent\nand adds links where missing.\n\nAlso moved sync/async and bulk create sections lower to\ngroup all the filtering, pagination and sorting sections\ntogether since they are inter-related.\n\n[0] https://review.opendev.org/c/openstack/neutron-lib/+/362877\n\nChange-Id: I895cc9aaee2546b95e93150c134f350a50346447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a44f91352a9ef7b5d414f94fdcbee6ef0eac621c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a44f91352a9ef7b5d414f94fdcbee6ef0eac621c"}]},"branch":"refs/heads/master"},"c28b3ab8fe5674887b643ac76dec2ed9837313d3":{"kind":"REWORK","_number":3,"created":"2024-04-04 23:00:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/914826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/914826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/914826/3"}}},"commit":{"parents":[{"commit":"b4cc4bb090fb54de837546972f57bad5644daf67","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b4cc4bb090fb54de837546972f57bad5644daf67"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-01 15:17:46.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-04 23:00:17.000000000","tz":-240},"subject":"Fix incorrect links for the filtering spec","message":"Fix incorrect links for the filtering spec\n\nThe spec networking_general_api_information.html was moved\nto api-ref in [0] but some of the links pointing to it\nwere not updated. This change makes all the refs consistent\nand adds links where missing.\n\nAlso moved sync/async and bulk create sections lower to\ngroup all the filtering, pagination and sorting sections\ntogether since they are inter-related.\n\n[0] https://review.opendev.org/c/openstack/neutron-lib/+/362877\n\nChange-Id: I895cc9aaee2546b95e93150c134f350a50346447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c28b3ab8fe5674887b643ac76dec2ed9837313d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c28b3ab8fe5674887b643ac76dec2ed9837313d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
