)]}'
{"id":"openstack%2Fneutron-lib~917263","triplet_id":"openstack%2Fneutron-lib~master~I967032c1dea690cd9da95ad3a8e382671186d971","project":"openstack/neutron-lib","branch":"master","topic":"netaddr1","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-05-30 21:24:22.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-05-30 18:41:02.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I967032c1dea690cd9da95ad3a8e382671186d971","subject":"Catch all exceptions in convert_ip_to_canonical_format","status":"MERGED","created":"2024-04-27 03:10:44.000000000","updated":"2024-05-30 21:25:34.000000000","submitted":"2024-05-30 21:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"917263-netaddr1","meta_rev_id":"e19b0d263b0a590c43f1dfccce9deff757f1f702","_number":917263,"virtual_id_number":917263,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-30 21:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-05-29 10:53:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-05-30 18:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-05-30 18:41:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":1,"date":"2024-05-04 00:45:00.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-27 05:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-29 20:04:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-05-01 19:22:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-05-29 10:53:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7c89697d3e1a85fd2729837161dd40f7a0a5d30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-27 03:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44599424908c9a7de4d1ef263438d66961bc1eb1","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-27 03:10:57.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7724b3c2f71e0745a21ffaf83e942ba7f3150192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 05:11:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03d5aaf343ed4c96900f009aa0ac16a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0306533fd84e445ea8b472ea008357d0 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7387dca14f94c93b917f397dec6e21b : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07456ef7d7854e4f8a07e5c3d0c7938e : FAILURE in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/650b5733d8ba4f888d4176b5884023a3 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ef6a69ce69941a5a9d9337a91d6052a : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf4a1da28adb4d5189373d65032a4666 : SUCCESS in 7m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d902228f2f440dab8a8e2f753e23ac8 : SUCCESS in 1h 09m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4212f21a84e04cfd8d92b1a3553cb02b : SUCCESS in 1h 42m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f142ec257a04e3692970e59f43b176e : SUCCESS in 1h 59m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c63e835f996a41738c138d591f8b5f0c : SUCCESS in 1h 08m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/77877ec601e4441ea2da87ed16c85537 : SUCCESS in 1h 51m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d1c720724a543528b3241062d3b9de8 : SUCCESS in 34m 13s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ea936f6f21be4321a82e8cbb71864350 : SUCCESS in 6m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b48321334e784aa0bb001088ef0d04e5 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cbed146f82321e9289b0fd5b9df00f2e53e4dbff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-30 17:29:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a5f068122aadaadae3273f22cdcd1ac44657346e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-30 17:30:23.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"ccdeb39547a8a4c40d6aa4ffec74bf3ddd24117f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 19:30:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2304cd281174d11822ef6cafddd9ea4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04616dac064744df8cda9409821064ef : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57c3915ed78f4cffb839dd07dfa48ede : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9d576872754cae93f503262adbaaab : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/578de96ddf1040de8abacdfab5daa8f7 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/877f2015fa444688b37cee9493598446 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e335b469cc0e422995378c3e1ff22f64 : SUCCESS in 6m 46s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb7095e3ee7a4659ad82bdbeb7578070 : SUCCESS in 1h 34m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/30780777c5064f11af4f6fb97ac217d2 : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc1a6080310a460295d788c32bfaee33 : SUCCESS in 1h 58m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3d7de93c203241988976d05d694fdf06 : POST_FAILURE in 1h 56m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/201cd0b5dc6a4990ae92dceadd82b3bd : SUCCESS in 1h 40m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb4a8ae05da9495ba85401eeff4736da : SUCCESS in 41m 40s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/041dff4fa4844cfba33678c4217927e7 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1ac6616ec5848ae9e5c4f0011684ab8 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5dff2caf8f316c7b55b1cb4c66368e9228d574f3","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 01:49:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ace0750a01aac8d551c0b9a0fe8bd0b00972401c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 04:19:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9155e988d3824d18b78ca4d18a17170e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810a00482eec4b5cb1274b528bacf747 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7972c631cc24d1f80bef63a5086285d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34edc342d76d42db866424d2cbe16332 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b41007053ae0417db9139ce77129e4d4 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d756aa0f4ff48818188521a9150e6a1 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe3a673fc83446ada68f4f87475ace5a : SUCCESS in 6m 48s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/707932168a72486e9a760f610eeaa884 : SUCCESS in 1h 30m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c8c0f24e4eae4b72be6ad6066879ce1f : SUCCESS in 1h 36m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f9e102faa1e4feba951cce71484e966 : SUCCESS in 1h 39m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1a6d2741b15d4075b28b8ed5bf4939aa : SUCCESS in 2h 01m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f47ddd32b5a54ff99af8ee25ce96b9a1 : SUCCESS in 1h 31m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c4e089082794d489dbaac2ebfd8c458 : SUCCESS in 33m 56s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c7d33256fae24ff9bb0e41002949379a : SUCCESS in 7m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/263f86374f78480e805da7135002a75f : SUCCESS in 42m 34s","accounts_in_message":[],"_revision_number":2},{"id":"17c4c3be952627025058458082195e5359a4ec6b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-01 16:23:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3797006843bfd63a268e064a4f17708a7624d662","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 18:14:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad466d2f404edbdf5dc9a306dcbd438724ab032c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 18:16:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b68597b17d9bd43fdef05741fe63571572ced509","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 18:17:18.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c28f74de6d237cc7ae671dec2895979e5b1e312","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 18:21:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4542a89ca6ef78692c682cb61aacb17bfe30ce7d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 18:24:13.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"368bc4289257834652e8886fe841292993699056","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-01 19:22:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f6ab23154e71104dbd5f8508e4096835db5dc2f","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 19:29:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47ab91700c080b4ea3ddfaf6818ce1bee55a72ac","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 19:31:32.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf8f584b0e6ea92cba8a4e133b216b6be42bb029","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-01 19:44:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48c6f1fbc90b3d0fb3efdb87c5dde5dfcc083c2b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 22:37:10.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96cfd6e3f06b5cbde7edff80e32612ed00c8375e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 23:06:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f03d19319f11f736a4ea47360209e4eb3d87c27f","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 23:06:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e059ea81fcded6b7eca2973bb9754af65f44c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 00:39:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a11f5b2bc2d4bc2b619d050a419f2aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672554ea6dce493999f70a17728c8407 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3754dc0122644f64a5110beda06feac6 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599aa24f910e430fa6f7a0716968ae59 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/754d2199b8ec4d628cede0bbed99ce88 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8d22ec893824b8893522b6af776056e : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/304916123eea4c84bb41d54deea5ae26 : SUCCESS in 6m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f79fdd57159c44359a1d00b4cc3223e9 : SUCCESS in 1h 09m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/75602bda3ad9419389e05d1865369750 : FAILURE in 1h 23m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa6ef57bc6864da6af3f614edbcb1a5b : SUCCESS in 1h 28m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4ff2059e4bb1493bbc74261084bf61a9 : SUCCESS in 1h 16m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/add51cfa9a7e4df0a502b386527e5174 : SUCCESS in 1h 31m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09eb8548191c4b77ac18c27d5437f920 : SUCCESS in 34m 16s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ab8d0946d0ed4e9ca007a2df2bd0dbe6 : SUCCESS in 5m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b839224308ad480f9faea1e394293487 : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":3},{"id":"0c883a6cd9624b1f140f3cad16e397df078be2e8","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-04 00:45:00.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"c5b7db53c768b5408314f956ba5a19b9d43cfc1f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-06 23:49:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"433813e04c4f285a4ba70157d3aae52f597f8a76","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-07 14:13:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80c4b1d129642d03bb33bee8741ff0888cd89b38","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 14:12:07.000000000","message":"Topic set to netaddr1","accounts_in_message":[],"_revision_number":3},{"id":"c9d656311fbc1afa20a22c989fb203ddd00ec2ae","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 16:41:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"afa3a51dd34b57c3bb33d38fa36e81ef0182ed9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 18:39:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da4bb4d7e05a4f81a598d59dc7b2a126\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8001111a9c5b4a658cbfbee95ac9e2c2 : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea6e577a02b243ab87af053146952794 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25c36d858f34b029171665ba2f6ee88 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceaa965527d64c68b3fccecc2a32f9c7 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/812bd1e2218f449f94b3a9ec5a626fd4 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3f0469a066749ae90d78d5b4652ebf9 : SUCCESS in 9m 50s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64ec6419318a4ec8988ef327d3544517 : SUCCESS in 1h 49m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4898ac1ce32647caa786139dd1cfa703 : SUCCESS in 1h 28m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d4843e038cb46bf87f56fe4a64ac160 : SUCCESS in 1h 39m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/273587ac99ac4c0c9d0fa560f297150f : SUCCESS in 1h 55m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac7fb780100347afbf502ccff1edfa98 : SUCCESS in 1h 50m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57ef65b9ae6448b0bc1463e95b792b6c : SUCCESS in 45m 43s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/47f8f533b5064390818e51f3116bfae7 : SUCCESS in 5m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b4b3986cb274941a846b18250e75e00 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":3},{"id":"69bc3e50206b00783f1e759df53661c89998f462","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-29 10:53:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b498c1f4ad8482244bf4db1425124e09828a1a17","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-30 18:41:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"73e0d7ff9cbd56fbc402ad2ef953ebd7550a0744","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 18:41:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"909a85adfe2c85470f19a14373c42d849555c8ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 21:24:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8651b33a58d5485aaeeafeb2e40a7c62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde711fd8a4741bcb9df1057bda4c2e6 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5587d141f18149d8af04435aab9b0245 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cac61062564c46f8be21d3ceeac3b45f : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82d7285a80b24b64b41349212f0ed741 : SUCCESS in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/675fcb1d27d34c99ab9067f5eb6e4fca : SUCCESS in 1h 40m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/25c22732814449fbad3ebd69fc37696f : SUCCESS in 1h 54m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27b90254e50842d9b07c45f0e5f98316 : SUCCESS in 2h 05m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4c96fd506634521b42eb4eb08440561 : SUCCESS in 2h 41m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/85393e7edddf4169ae516a415e9a818d : SUCCESS in 2h 23m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de2eff693b294830862e5941f14e4788 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":3},{"id":"343baaa2c7176d649f965e82f3fa34165af9c2d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 21:24:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e19b0d263b0a590c43f1dfccce9deff757f1f702","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 21:25:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1588117e1aa940f4a256734a8d63ebdb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4b73d72fd64d71a99846e661947244 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1020da3eacff914259b3411f30ecb564d9f047c","revisions":{"db90a0c0c8d9ccd8d32f8cfc0cac38481322ea45":{"kind":"REWORK","_number":1,"created":"2024-04-27 03:10:44.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/63/917263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/63/917263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/1"}}},"commit":{"parents":[{"commit":"688b135d86188848e214bd2ac18e92f9120b4be9","subject":"Merge \"Fix incorrect links for the filtering spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/688b135d86188848e214bd2ac18e92f9120b4be9"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 03:05:47.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 03:10:32.000000000","tz":-240},"subject":"WIP: catch-all exceptions in convert_ip_to_canonical_format","message":"WIP: catch-all exceptions in convert_ip_to_canonical_format\n\nRecent netaddr started to raise AttributeError for subnetpools values\nlike [{\u0027start\u0027: ..., \u0027end\u0027: ...}].\n\nThe converter is meant to be graceful towards inputs that are not proper\nip addresses.\n\nInstead of adding another exception type to the list and hope that the\nlist of expected types covers everything netaddr may raise, this patch\nmakes the converter catch all exceptions.\n\nTODO: add a test case.\n\nChange-Id: I967032c1dea690cd9da95ad3a8e382671186d971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/db90a0c0c8d9ccd8d32f8cfc0cac38481322ea45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/db90a0c0c8d9ccd8d32f8cfc0cac38481322ea45"}]},"branch":"refs/heads/master"},"9390829b87ca8ebdfec43717f248dd59da0364b0":{"kind":"REWORK","_number":2,"created":"2024-04-30 17:29:55.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/63/917263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/63/917263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/2"}}},"commit":{"parents":[{"commit":"688b135d86188848e214bd2ac18e92f9120b4be9","subject":"Merge \"Fix incorrect links for the filtering spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/688b135d86188848e214bd2ac18e92f9120b4be9"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 03:05:47.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-30 17:29:43.000000000","tz":-240},"subject":"Catch all exceptions in convert_ip_to_canonical_format","message":"Catch all exceptions in convert_ip_to_canonical_format\n\nRecent netaddr started to raise AttributeError for subnetpools values\nlike [{\u0027start\u0027: ..., \u0027end\u0027: ...}].\n\nThe converter is meant to be graceful towards inputs that are not proper\nip addresses.\n\nInstead of adding another exception type to the list and hope that the\nlist of expected types covers everything netaddr may raise, this patch\nmakes the converter catch all exceptions.\n\nChange-Id: I967032c1dea690cd9da95ad3a8e382671186d971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9390829b87ca8ebdfec43717f248dd59da0364b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9390829b87ca8ebdfec43717f248dd59da0364b0"}]},"branch":"refs/heads/master"},"d1020da3eacff914259b3411f30ecb564d9f047c":{"kind":"REWORK","_number":3,"created":"2024-05-03 23:06:30.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/63/917263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/63/917263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/63/917263/3"}}},"commit":{"parents":[{"commit":"577d72384faccbd8203b16c35a12ba4dfec3ccca","subject":"Merge \"pep8: enforce test file names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/577d72384faccbd8203b16c35a12ba4dfec3ccca"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 03:05:47.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 22:56:14.000000000","tz":-240},"subject":"Catch all exceptions in convert_ip_to_canonical_format","message":"Catch all exceptions in convert_ip_to_canonical_format\n\nRecent netaddr started to raise AttributeError for subnetpools values\nlike [{\u0027start\u0027: ..., \u0027end\u0027: ...}].\n\nThe converter is meant to be graceful towards inputs that are not proper\nip addresses.\n\nInstead of adding another exception type to the list and hope that the\nlist of expected types covers everything netaddr may raise, this patch\nmakes the converter catch all exceptions.\n\nChange-Id: I967032c1dea690cd9da95ad3a8e382671186d971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d1020da3eacff914259b3411f30ecb564d9f047c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d1020da3eacff914259b3411f30ecb564d9f047c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
