)]}'
{"id":"openstack%2Fneutron-lib~917855","triplet_id":"openstack%2Fneutron-lib~master~I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d","project":"openstack/neutron-lib","branch":"master","topic":"netaddr1","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-06-05 21:47:06.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-06-05 13:29:11.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-06-05 19:08:01.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d","subject":"tests: return same message for validate_[subnet*|route_cidr]","status":"MERGED","created":"2024-05-01 15:39:15.000000000","updated":"2024-06-05 21:48:06.000000000","submitted":"2024-06-05 21:47:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"917855-netaddr1","meta_rev_id":"1995b37262b6f653cc2669a3d6748288d0039973","_number":917855,"virtual_id_number":917855,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 21:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-06-05 13:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-06-05 19:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-06-05 19:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 15:42:22.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-05-01 18:25:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-30 19:25:15.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"6b2eb06cbadfcebc34ac8aee936d346ca0303f7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 15:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11494696943645250c26d2af27edb1f77d6310c4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 15:39:23.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"4840dcdc5933cab53b0813565ff5a5947b9db27e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 15:41:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33dfe381eb3ee68c5931fdbdd23fa25cea3817a5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 15:42:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dffb334e50d74f2dc832f5406271ca0dcfc3e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 15:55:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"27429b8e719150e582dfb2b2c0841ecddc200a32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 16:04:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3a9d0e1be84b03def733e5c33e0821eee6a1f224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 18:25:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991fe3f390e94b31b9bfc372abbb9ff4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44cccf0bfd2e459aa98cfbfc85b2c122 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f2d16b3f604414095819d873f3b4167 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6771b0277a2747fb96b2db718ec45443 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbdcb331996343bcab2188b3104fe835 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f52e182cc5b7490cbc38bdd3911e0ad1 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1882db7f864e4c50bda8d23324bf1483 : SUCCESS in 6m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/168c5fb58cd74f80aa637a1d418766bf : SUCCESS in 1h 16m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b3e1f4ddb3204032914ab5af4099c0af : SUCCESS in 1h 37m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89dcd7bc9706440bbbb2e39f8ca954e0 : SUCCESS in 2h 07m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a4e26c265d074d6694c8e862f40b92ff : SUCCESS in 1h 24m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/373130b27541452a943f5bb4636255d4 : SUCCESS in 2h 18m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/deb4eebf64864f939a8f789d394257ec : SUCCESS in 42m 04s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7bdd1ea45c7444628bfde8b5a4adce75 : SUCCESS in 7m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/284ff632e57f4426af7d8646c769efc5 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":3},{"id":"90e3401f9603bd6bf555e956848d11bb38a6a29e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 19:34:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e895e89257ff0768f2da5e1cd72bf75d75bf99f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 19:43:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"740089a2892810569a3ed99a670717a759dbfaf6","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-01 21:31:52.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"f802eca09aa16a11284a77288d42926ce9198485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 21:38:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c69ca9ca37f42d4aa2508e3fa481d86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44ef66a3f584686b63790a42169f954 : SUCCESS in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88c6cea70a294186b59e3bd701da4946 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed672c976e5477ba5c69e10a538bb7c : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c305fe0c7c0e4b908c304ae0cb0014dd : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e727eafcf5141ad89abf1c8f3860065 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9132d1d4f7974bafaf90dac22fab3203 : SUCCESS in 7m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76a8f148cdd44c2ab71b03a4e69f75b3 : SUCCESS in 1h 30m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0c5eb8982a6a4d9493a01a51a4963ef9 : FAILURE in 1h 33m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42e10723ae534672ac4cd165101c511f : SUCCESS in 1h 51m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9e66f3dc45504046a7e3688f44714689 : SUCCESS in 1h 26m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a42cf0b3e4ee45d9aa7853925272fd3a : SUCCESS in 1h 44m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fee905307f8f4166a8c20b1e8c058a8a : SUCCESS in 46m 15s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a047a5970a77447f9b682cb0f15249e6 : SUCCESS in 4m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c873f489a48d49a7bd36f4601b4fb8fb : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":4},{"id":"0c0173bf9db68f6f4b13df9c38944c18711e5c46","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 15:38:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e25f499f6788e743f2b1d3c8fb22432aaef1d4d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 17:48:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cae1bf1053f0497d9bc28550528a055d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f589e648a754bd8b808b9aba6a8ece4 : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bfa71443f1c4a9082bacf1d256024aa : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef0c60ef9054f0c9988eb4fc9cb79ee : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/773edcf37931406db8ad43fc8c4c9168 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a91086403424809ba93657afc30ae24 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/210497d4429249b08e45c2ab1620af65 : SUCCESS in 8m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4e62fb37a924983ab2461ce40e0c625 : SUCCESS in 1h 31m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/810ae7d7653e4a11b1641110a7396021 : SUCCESS in 1h 41m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fc839467fc8548f08c68af5636101dfa : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8ceb37f1ecd3482484d301a56cf4cd84 : SUCCESS in 2h 02m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2ce0d89c2ab64f54b33544b68e6c3983 : SUCCESS in 1h 49m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4a83ffed20aa49cdb300eb324aeaf610 : SUCCESS in 42m 22s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/230a477471224354bd6bf2c704fa2eeb : SUCCESS in 4m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff754c05147645c2865b96054db54a76 : POST_FAILURE in 1h 08m 16s","accounts_in_message":[],"_revision_number":4},{"id":"608d8866b9c6aafc0fea21feb7930eeb55ad8e4c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 19:23:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9041eb235a83f64786ee270d37332e3241ff1dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:57:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8463492c7234a49baf594a1a2f8e56c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac3eddfe2b34e2c86735255064ce2c3 : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccd0713c804c4f31856eee22bc2a0460 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8cf43018c0c45ec9d536ff08b5ab9f4 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4e413a592b42c896358c4b8902e37f : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69e9e12587744bb19735511dec1bcdc0 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f926a0f69064e128d3577511a5ea789 : SUCCESS in 7m 30s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17ec7b3ab33c4ff8a412af2de7fe2f15 : SUCCESS in 1h 18m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/957ba3bec880465d9e9d6ff32f83b00e : SUCCESS in 1h 17m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ddf8e81e9f57429f8a5ad6bac3ad387a : SUCCESS in 1h 31m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/059344590ad945ecb11d65f207ab03c9 : SUCCESS in 1h 13m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed5829b4432d4a11a4ff3e63d3c34e87 : SUCCESS in 1h 26m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/471b81f301f842448570c80607bf9502 : SUCCESS in 34m 35s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2006888ff6584100b2bb6c1ac591bd5b : SUCCESS in 5m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f29b4819cfce4931a960da4a36854f59 : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":4},{"id":"2f876b68cea6217f8adade9451f9a6e7ac52dfd7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 14:12:14.000000000","message":"Topic set to netaddr1","accounts_in_message":[],"_revision_number":4},{"id":"8118bf8c57ea1f6521b4b179fd77924ba738986f","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 14:31:51.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"869502617c78f900adbcbf37c477633380c4b97d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 14:32:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f65ba5c3a839dd04d327d9688eeb54a9e78fba62","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 14:33:06.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"94eef214909f1b350fd1680428deb8b9e9b10aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 16:46:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4786e4cb92614c06b0660d6d41781ebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4b9db6c5d345df8e9cd58e20fb8a97 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4214011726804181afe64e36929abfaa : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f31928aa83446569cc2127698fdd990 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bda9eaef18ec42cb87a55ac594a73bfb : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73bf1f04d7534ba2b0e5004f662b57f9 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edbfceb082fe455b8a4aebe11c9e377a : SUCCESS in 6m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61a3a31f60414a699e578df22ec63976 : SUCCESS in 1h 34m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f92a18dcdc544ee9addaaa8c065d786c : SUCCESS in 1h 40m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34934cc423a04559948cbc3be635ab61 : SUCCESS in 1h 44m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/430adfde15cc4432ac9f262f0eab7ea5 : SUCCESS in 2h 05m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0604d80dcb04716aa08723ce5f59cd6 : SUCCESS in 1h 28m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61cc4a70516d4273a11af534bbcfe23a : SUCCESS in 31m 51s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a97e0cd3998b4722ae1abe25a4ad8c82 : SUCCESS in 6m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6d65a693f1545d7a318cdaddcf4a533 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":5},{"id":"5f7a6aabb68cf77d15fa2efe064f65064fbd7620","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-06-05 13:29:11.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a63cbb57c3db7b541f8def8d4d184a53bd3ae76a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-05 19:08:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c03b74c8d9c5f8d06fa19491203dcd46ac689780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 19:08:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6666824cc07df3c9f7c025ccc323323fbd6f7900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 21:47:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77786d458aba4f9092edb8e429388f4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca18dd39370a4f89aedf56cd6a725eba : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e7b5a18f704c169fbe028b7331ed51 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de016fc9a7be4c1cbd27d149148ca039 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7d27e0cc93942d09694102a7904dd8c : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2eab421a3926480faf2fa11ba3ea98cc : SUCCESS in 1h 52m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba9a5b434bb14c9d9a5d4bdb2873b241 : SUCCESS in 1h 55m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bf254f4b849245fbb5c08db5d154e2e0 : SUCCESS in 1h 58m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/385bf6d9bd024bef9c54612819ef785a : SUCCESS in 2h 37m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4b51831385c7499389da15d85c52fc64 : SUCCESS in 1h 58m 34s","accounts_in_message":[],"_revision_number":5},{"id":"d19502c70932e27d21c59ef21b0fd4e943286a29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 21:47:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1995b37262b6f653cc2669a3d6748288d0039973","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 21:48:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d386eeab56794926b2956e4317bb865e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8487cb11e136469aaf5b370ed42af597 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"37e915ea1c56f2289dc911b813e317e3dd1e6dc4","revisions":{"b1144fa2a624df86a80f5fd54253d3618baa9b99":{"kind":"REWORK","_number":1,"created":"2024-05-01 15:39:15.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/55/917855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/55/917855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/1"}}},"commit":{"parents":[{"commit":"4d5e0d7a904d54164b1d328b5bd15e03a6cb47aa","subject":"Rename test_conversions -\u003e test_converters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d5e0d7a904d54164b1d328b5bd15e03a6cb47aa"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:36:16.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:36:16.000000000","tz":-240},"subject":"tests: return same message for validate_[subnet*|route_cidr]","message":"tests: return same message for validate_[subnet*|route_cidr]\n\nIn netaddr 1.0.0, more requirements are enforced for input cidrs. This\nmakes the library raise exceptions for some values that were accepted\nbefore and that we then manually validated (e.g. for the fact that the\nlibrary normalized value is identical to input.)\n\nThis patch makes the validators behave identically for both netaddr\nversions (returning the same messages.)\n\nThis fixes some test failures with netaddr 1.2.1.\n\nChange-Id: I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1144fa2a624df86a80f5fd54253d3618baa9b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1144fa2a624df86a80f5fd54253d3618baa9b99"}]},"branch":"refs/heads/master"},"16b2064dae0513aab5468518a6e4efee1d26f2b6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-01 15:55:49.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/55/917855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/55/917855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/2"}}},"commit":{"parents":[{"commit":"9390829b87ca8ebdfec43717f248dd59da0364b0","subject":"Catch all exceptions in convert_ip_to_canonical_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9390829b87ca8ebdfec43717f248dd59da0364b0"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:36:16.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:54:44.000000000","tz":-240},"subject":"tests: return same message for validate_[subnet*|route_cidr]","message":"tests: return same message for validate_[subnet*|route_cidr]\n\nIn netaddr 1.0.0, more requirements are enforced for input cidrs. This\nmakes the library raise exceptions for some values that were accepted\nbefore and that we then manually validated (e.g. for the fact that the\nlibrary normalized value is identical to input.)\n\nThis patch makes the validators behave identically for both netaddr\nversions (returning the same messages.)\n\nThis fixes some test failures with netaddr 1.2.1.\n\nChange-Id: I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/16b2064dae0513aab5468518a6e4efee1d26f2b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/16b2064dae0513aab5468518a6e4efee1d26f2b6"}]},"branch":"refs/heads/master"},"1d3b11c148fec68e9ea06d8103c7d5b7578c0aa4":{"kind":"REWORK","_number":3,"created":"2024-05-01 16:04:46.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/55/917855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/55/917855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/3"}}},"commit":{"parents":[{"commit":"9390829b87ca8ebdfec43717f248dd59da0364b0","subject":"Catch all exceptions in convert_ip_to_canonical_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9390829b87ca8ebdfec43717f248dd59da0364b0"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:36:16.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 16:04:35.000000000","tz":-240},"subject":"tests: return same message for validate_[subnet*|route_cidr]","message":"tests: return same message for validate_[subnet*|route_cidr]\n\nIn netaddr 1.0.0, more requirements are enforced for input cidrs. This\nmakes the library raise exceptions for some values that were accepted\nbefore and that we then manually validated (e.g. for the fact that the\nlibrary normalized value is identical to input.)\n\nThis patch makes the validators behave identically for both netaddr\nversions (returning the same messages.)\n\nThis fixes some test failures with netaddr 1.2.1.\n\nChange-Id: I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d3b11c148fec68e9ea06d8103c7d5b7578c0aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d3b11c148fec68e9ea06d8103c7d5b7578c0aa4"}]},"branch":"refs/heads/master"},"91ecce40f21fa89f7f89831702ac7d1cb4559b1e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-01 19:43:37.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/55/917855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/55/917855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/4"}}},"commit":{"parents":[{"commit":"688b135d86188848e214bd2ac18e92f9120b4be9","subject":"Merge \"Fix incorrect links for the filtering spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/688b135d86188848e214bd2ac18e92f9120b4be9"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:36:16.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 19:43:29.000000000","tz":-240},"subject":"tests: return same message for validate_[subnet*|route_cidr]","message":"tests: return same message for validate_[subnet*|route_cidr]\n\nIn netaddr 1.0.0, more requirements are enforced for input cidrs. This\nmakes the library raise exceptions for some values that were accepted\nbefore and that we then manually validated (e.g. for the fact that the\nlibrary normalized value is identical to input.)\n\nThis patch makes the validators behave identically for both netaddr\nversions (returning the same messages.)\n\nThis fixes some test failures with netaddr 1.2.1.\n\nChange-Id: I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/91ecce40f21fa89f7f89831702ac7d1cb4559b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/91ecce40f21fa89f7f89831702ac7d1cb4559b1e"}]},"branch":"refs/heads/master"},"37e915ea1c56f2289dc911b813e317e3dd1e6dc4":{"kind":"REWORK","_number":5,"created":"2024-05-14 14:32:52.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/55/917855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/55/917855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/55/917855/5"}}},"commit":{"parents":[{"commit":"688b135d86188848e214bd2ac18e92f9120b4be9","subject":"Merge \"Fix incorrect links for the filtering spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/688b135d86188848e214bd2ac18e92f9120b4be9"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-01 15:36:16.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-14 14:32:45.000000000","tz":-240},"subject":"tests: return same message for validate_[subnet*|route_cidr]","message":"tests: return same message for validate_[subnet*|route_cidr]\n\nIn netaddr 1.0.0, more requirements are enforced for input cidrs. This\nmakes the library raise exceptions for some values that were accepted\nbefore and that we then manually validated (e.g. for the fact that the\nlibrary normalized value is identical to input.)\n\nThis patch makes the validators behave identically for both netaddr\nversions (returning the same messages.)\n\nThis fixes some test failures with netaddr 1.2.1.\n\nChange-Id: I4bc9c4238ed26cdc7ebf6e0c877ea4bd9c79c36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37e915ea1c56f2289dc911b813e317e3dd1e6dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37e915ea1c56f2289dc911b813e317e3dd1e6dc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
