)]}'
{"id":"openstack%2Fneutron-lib~918149","triplet_id":"openstack%2Fneutron-lib~master~If768ab6a5f92d200841a734087bbc8fba8870dc6","project":"openstack/neutron-lib","branch":"master","topic":"netaddr1","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-05-30 18:00:48.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-30 15:51:48.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-05-29 17:08:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If768ab6a5f92d200841a734087bbc8fba8870dc6","subject":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field","status":"MERGED","created":"2024-05-03 23:06:30.000000000","updated":"2024-05-30 18:01:46.000000000","submitted":"2024-05-30 18:00:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918149-netaddr1","meta_rev_id":"bb623645de2f227ad662edc258efd5cc382aaab4","_number":918149,"virtual_id_number":918149,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-30 18:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-05-30 15:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-05-29 17:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-05-30 15:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-04 00:53:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 23:30:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-05-27 11:26:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-29 17:08:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"bf57ec6c4940c7588e0cf22a090cb81c9247f0c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 23:06:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55abd25f12d3cb27d268cb03f8238d0212e3c190","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 23:07:37.000000000","message":"Patch Set 1: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76194cabc202142655990feffb9aeba826946173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 00:53:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea992007a5c14a04b779be6edef44e3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96285ee069604b08aa123d865930ba85 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d955f5953d44400ba371c8851d01e50 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e91f742aa446b29ee889a4010b5e26 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9ed9f2a46324a8b8ca8f51e5bdb956b : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57212b33cfb34084a5beab9e1ca4a864 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c87b493ea1c4f738fdc44e2683e9c6a : SUCCESS in 8m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67106dea48cc4140bca259780e442175 : SUCCESS in 1h 19m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc9b5da8059e4bffad12d08072958079 : SUCCESS in 6m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ce7a3792d6ba4ec791017165e3e4db74 : SUCCESS in 1h 39m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2353a798e1354e06a75c3f84d51e71a1 : SUCCESS in 1h 34m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/699ea6cdc09b4a8da7fa8a2a6c802b19 : FAILURE in 1h 14m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f5c17c73060c426f993201d87906ecae : SUCCESS in 1h 44m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21f9879627ec4f7eab555eb747005be6 : SUCCESS in 30m 31s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/eba364a9ba9a495683cde9ab324f3704 : SUCCESS in 5m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd177cba4adb42f7b7549cdb10966a29 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ef6cefea6dfcf5b6fa7716f51a27626d3294a4b8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-06 23:30:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"74823136e68cee9f756e1bc106b3152d6a9d541a","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-07 14:14:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5d5e61495d973482de81a1dcb05914498a7ccfb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 01:03:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b268ade8de1c7a80fac9b60f255dc78d8c21d03e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 02:51:35.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1df641d7afa04689b54747428d6c7b7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb9c537083e4a98ace857390800d06b : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c838c9b4df484386a13bc8626f575cd6 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6727b24493774d5893f961f3dd8fd845 : FAILURE in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c94799773a4dfab8ace0fe553535a4 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08c5013b2e1e4f1fb9e4c1df8b843e19 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2edb9d60d95d40ae9631db8fd2c9139f : SUCCESS in 8m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a92b3d7fecc464fb1948053fffbcc19 : SUCCESS in 1h 29m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4634ab808fc143098c0a2469a8716cb9 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/78985b367d854539ba0d10b42e5ef93d : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93498f3328664f0eb46b2da09e21ebcb : SUCCESS in 1h 44m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7b5d243ef37b4c23a0fe0a4007709160 : SUCCESS in 1h 41m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c06dbef429a74526af86c3b05317b144 : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9eea80b225284b16beff48f70f4f95a0 : SUCCESS in 38m 43s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3db26a60452d4383a529f5934d33b069 : SUCCESS in 5m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62a45cfbf3434772a9f84e65e3d414d0 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":2},{"id":"79b5f105f3ffd5a274aa82a758887f67b6a0afaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 13:09:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2f7b196d42e6dc4c194764e51f7666a48f493cbf","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 13:10:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9483639d73c7c66e7220fb6421da76772f4da9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 13:34:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b2c87ad762a3e4b47c3aed69378b00b01d46c359","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 14:12:10.000000000","message":"Topic set to netaddr1","accounts_in_message":[],"_revision_number":4},{"id":"caaac5656d4d553201669df43012b7aa4cc8de53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:31:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e8f3aff7e504632953d25d045026e8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0795967307d4ebda71e5fe85ab7e718 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dcc684b96af46f9842aa7e832794d8f : POST_FAILURE in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c183b162f53f4420963d1f1185d75997 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe74d924229f46639c0722c6a911f4b6 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eead06e558b6481281388b9fa93eccdb : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/278fde6912da493cbd3b430c3484eda1 : SUCCESS in 7m 55s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/915e01bb779c4c9681cd21aa19537a07 : SUCCESS in 1h 50m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d44cf09b41384dc18d1eaea286198395 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ce11502e09d463e9e8e452073346365 : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1319aebe94fc4ca3af0956ce5ce88fed : SUCCESS in 1h 38m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b20711fbb487488c93cc0a5b397e2890 : SUCCESS in 1h 55m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ff5fd43e68804b18b0f81ad0b8addec9 : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/afa6ad97bcaf4dcbad1f8e9f9c762acb : SUCCESS in 36m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/53c7b054ba26420ebc8fa45b5e417724 : SUCCESS in 5m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a0bde0617584f78b3d9447f8e89db2a : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":4},{"id":"845077b8d48a358d798d344a40a37634a73fc9ba","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 16:31:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78f787f82c709031b4473f7b29649e6024e0383d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 18:30:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2af094b5de47414cb78265c779779169\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01251d4ba3434d9492c1ec0f0ccd7299 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec0814a856f143ae8cc2fae8913426ef : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aab777fcf1f4050a7017902fc1ea440 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71b8076431d2496fbed398ba55afaa74 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7bad3db6a764e17a4cbdcf2b3619d67 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db7c818d4a0a479e8c0596d1931b82e7 : SUCCESS in 7m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8af743053674393a0790a15209a9fe9 : SUCCESS in 1h 37m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a11d7e1a8a641249cd93e2ea45e799d : SUCCESS in 3m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9e9fc8d7d82640fa82815ab13b2f7df2 : SUCCESS in 1h 47m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be973670555245cfa3dcfe6a7421235e : SUCCESS in 1h 54m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/47db798cb0e04fd0a1605484b9daf4ae : FAILURE in 1h 58m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cb6a04d74bde41a5a55f326e5eb903ed : SUCCESS in 1h 34m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c57728eca0c34894a92c0a2598be8a75 : SUCCESS in 35m 45s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9a07259dba254dc1988495704941a283 : SUCCESS in 6m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2657d2d29ee84720a6c6a268050d1684 : FAILURE in 48m 28s","accounts_in_message":[],"_revision_number":4},{"id":"9e58fad612cde91319d143ff2830d94f9a2514aa","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-15 21:01:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59a143872f3203683ef5f1eb5ffc10f0b8454f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 22:59:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970156d448e64d31a2c8be8bf077639f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6669a01bc9994ad6b9b98488b49d16c1 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71315103098e4ef684d586d8ab3673eb : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6d0fe3d8d24c92bdeb2e6b19f41008 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f046ed7fab5b4397b14123f706fb9a92 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/157d3339a331497192a3e9af3cc49cf3 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5774d4a6eb9f4a03b9e8c4d68b73f826 : SUCCESS in 6m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29124c01b60648d5a997394435b7e85e : SUCCESS in 1h 48m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62a6d0189c5c4b838613fa1658f80aa6 : SUCCESS in 3m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eaf790e52f8f4ef1af9500c1b1e70f83 : SUCCESS in 1h 07m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63798a64bad144028f40cbb86faadac1 : SUCCESS in 1h 27m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0a2d730190ba45248e824d5e5a8df9ee : SUCCESS in 1h 33m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bda132f3cd0a4b48bbfee5b4b24ad878 : SUCCESS in 1h 18m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/917b36c3117045949340dbc2e9f48e71 : SUCCESS in 31m 35s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6f75a4d8499d427c8090a567695500db : SUCCESS in 5m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff64001559724895acdcdbcbad6ac9c9 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":4},{"id":"53e58fd8488a751e371f746ab971b1bea5ddac2a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-27 11:26:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"167b99e68acf122606928099ef852fce3f3b6f23","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-29 02:34:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"28ce7a97db7dadea1c795d519dd11bff9a6658bc","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-29 15:10:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fb68b584b4198826367b1f027703064f32947893","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-29 15:25:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fe06e288fb2494dc1bd3166bc6e969b8ac80501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-29 16:24:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"10af79f49b911c0a0efd980d8f0e97edab6aa63a","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-29 16:24:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ce34b2ee8eb9030b2c8965a0d8faff8df951b2be","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-29 17:08:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"363cf706587acc2d1d28254bfe3f30a0323f7f6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 18:51:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa805d6794ca49a8bcd4d5f15beb0468\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5526bbe1222740b7961611ffd60b38b9 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21fba2f7db6343209bf284c1878585a9 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4295524fee944588f4b3268c3db5320 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72fe98396d814349ae656e9dc715b583 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61a6a07327864a33ad3bc5d5ec4330a9 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69c0216c203e4277abaa5eb425f0570d : SUCCESS in 6m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/225df51129ed489ea625a517ea7de1fa : SUCCESS in 1h 47m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e4adc5a1d048afbad43f6cc848408b : SUCCESS in 3m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d3ef4eeadcfe4463b3a455fbe253aa7c : SUCCESS in 2h 15m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6243d184b7d46639c78a99953c53d8a : SUCCESS in 2h 24m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/dfeb08bd63244067ba17f10faac61caa : SUCCESS in 2h 19m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a42cc0935c2b4a328319bcef30d3ad20 : SUCCESS in 2h 11m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c67a87efec8e474fbbc9607cf5713dd5 : SUCCESS in 32m 38s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7e1aa335ebbd49a0b72c6e71351d5eea : SUCCESS in 6m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9710f98675504824a6030cc67fcd3020 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":5},{"id":"8d8e06280cfe765627649bb2138fd91bbf38566b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-30 15:51:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c612a40b2814d58c3e700f48bb17fa8880ee2376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 15:52:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"17b087cc276e0be304474cebe3911b4bde7ca786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 18:00:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa84fe2a33b24e82ad31eb103f60bdd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f80adb349848f0ab3036346c07b96a : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db9880525faa4cceafa64211dc074fe2 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c09b98f9d72e40a9a073099e2a1ad830 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fd5fcf4458b452d9f147153afb2aa83 : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f9fa7697c5449b4b3e4e2d57bc2802f : SUCCESS in 1h 44m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca7b012931b54146834f7f5c0f27f27c : SUCCESS in 3m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3c3fd205f7e94dbeadb9d1f09971de23 : SUCCESS in 2h 06m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c45b96b336ae469595997751a8049ff2 : SUCCESS in 1h 55m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/099479c3a8d049a7ba1c38207317a209 : SUCCESS in 1h 46m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8452d0bc1df349f89d7e71efc208eb26 : SUCCESS in 1h 34m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d80f1a5685af483d896c788e7171a96c : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":5},{"id":"12de69b836f158c54a64e36ede918014aaa1865f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 18:00:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"bb623645de2f227ad662edc258efd5cc382aaab4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 18:01:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b2f6c13f0d94e65841ef9e726221d87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e884eef50041b7b66b11b9f0124a6d : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c36d06ccff046158f35ba9392515474 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e9275425db8c1315ad5efdd18d42f107e370768f","revisions":{"27ae4f8c106554d11fbc3c73ab34435f28a6ae6e":{"kind":"REWORK","_number":1,"created":"2024-05-03 23:06:30.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/918149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/918149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/1"}}},"commit":{"parents":[{"commit":"d1020da3eacff914259b3411f30ecb564d9f047c","subject":"Catch all exceptions in convert_ip_to_canonical_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d1020da3eacff914259b3411f30ecb564d9f047c"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 22:56:27.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 23:06:20.000000000","tz":-240},"subject":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field","message":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field\n\nBefore the patch, `start` and `end` addresses in `allocation_pools` were\nnever canonicalized because an invalid converter was used for this field\nthat assumed a string as an input format, and not a list of dicts with\n`start` and `end` keys, which is the correct format for the field.\n\nThe original invalid converter was added in\nIbef0bd9a2f7b7dd8ddccc183f17aa31e426c7fd0.\n\nRelated-Bug: #1531103\nChange-Id: If768ab6a5f92d200841a734087bbc8fba8870dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27ae4f8c106554d11fbc3c73ab34435f28a6ae6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27ae4f8c106554d11fbc3c73ab34435f28a6ae6e"}]},"branch":"refs/heads/master"},"ce5675e85822a72c43d085cd88d49de973254b76":{"kind":"REWORK","_number":2,"created":"2024-05-14 01:03:57.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/918149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/918149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/2"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 22:56:27.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-14 01:03:52.000000000","tz":-240},"subject":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field","message":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field\n\nBefore the patch, `start` and `end` addresses in `allocation_pools` were\nnever canonicalized because an invalid converter was used for this field\nthat assumed a string as an input format, and not a list of dicts with\n`start` and `end` keys, which is the correct format for the field.\n\nThe original invalid converter was added in\nIbef0bd9a2f7b7dd8ddccc183f17aa31e426c7fd0.\n\nRelated-Bug: #1531103\nChange-Id: If768ab6a5f92d200841a734087bbc8fba8870dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ce5675e85822a72c43d085cd88d49de973254b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ce5675e85822a72c43d085cd88d49de973254b76"}]},"branch":"refs/heads/master"},"447a1fa324402a87b35353aa8ae5caa9a4b8c7be":{"kind":"REWORK","_number":3,"created":"2024-05-14 13:09:24.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/918149/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/918149/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/3"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 22:56:27.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-14 13:09:19.000000000","tz":-240},"subject":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field","message":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field\n\nBefore the patch, `start` and `end` addresses in `allocation_pools` were\nnever canonicalized because an invalid converter was used for this field\nthat assumed a string as an input format, and not a list of dicts with\n`start` and `end` keys, which is the correct format for the field.\n\nThe original invalid converter was added in\nIbef0bd9a2f7b7dd8ddccc183f17aa31e426c7fd0.\n\nRelated-Bug: #1531103\nChange-Id: If768ab6a5f92d200841a734087bbc8fba8870dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/447a1fa324402a87b35353aa8ae5caa9a4b8c7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/447a1fa324402a87b35353aa8ae5caa9a4b8c7be"}]},"branch":"refs/heads/master"},"185ed36d8981ac4d8ad422c7ade77a8515a2483f":{"kind":"REWORK","_number":4,"created":"2024-05-14 13:34:10.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/918149/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/918149/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/4"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 22:56:27.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-14 13:34:04.000000000","tz":-240},"subject":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field","message":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field\n\nBefore the patch, `start` and `end` addresses in `allocation_pools` were\nnever canonicalized because an invalid converter was used for this field\nthat assumed a string as an input format, and not a list of dicts with\n`start` and `end` keys, which is the correct format for the field.\n\nThe original invalid converter was added in\nIbef0bd9a2f7b7dd8ddccc183f17aa31e426c7fd0.\n\nRelated-Bug: #1531103\nChange-Id: If768ab6a5f92d200841a734087bbc8fba8870dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/185ed36d8981ac4d8ad422c7ade77a8515a2483f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/185ed36d8981ac4d8ad422c7ade77a8515a2483f"}]},"branch":"refs/heads/master"},"e9275425db8c1315ad5efdd18d42f107e370768f":{"kind":"REWORK","_number":5,"created":"2024-05-29 16:24:26.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/918149/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/918149/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/918149/5"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 22:56:27.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-29 16:24:17.000000000","tz":-240},"subject":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field","message":"api: canonicalize ipv6 addresses in `subnet:allocation_pools` field\n\nBefore the patch, `start` and `end` addresses in `allocation_pools` were\nnever canonicalized because an invalid converter was used for this field\nthat assumed a string as an input format, and not a list of dicts with\n`start` and `end` keys, which is the correct format for the field.\n\nThe original invalid converter was added in\nIbef0bd9a2f7b7dd8ddccc183f17aa31e426c7fd0.\n\nRelated-Bug: #1531103\nChange-Id: If768ab6a5f92d200841a734087bbc8fba8870dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9275425db8c1315ad5efdd18d42f107e370768f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9275425db8c1315ad5efdd18d42f107e370768f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
