)]}'
{"id":"openstack%2Fneutron-lib~919757","triplet_id":"openstack%2Fneutron-lib~master~If935c4f2281938fb31e11758b7c0f6b504894833","project":"openstack/neutron-lib","branch":"master","topic":"type_tooling","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-11-12 00:34:45.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-09-11 17:15:16.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2024-11-12 00:34:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If935c4f2281938fb31e11758b7c0f6b504894833","subject":"[WIP] Add initial mypy tooling for type annotations","status":"ABANDONED","created":"2024-05-15 16:54:33.000000000","updated":"2024-11-12 00:34:45.000000000","total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"82d38395516447cd69d5c8c67248a849c282e80c","_number":919757,"virtual_id_number":919757,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-15 21:06:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":-1,"date":"2024-06-06 21:54:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 21:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 08:31:04.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-06-06 21:54:29.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-06-10 06:49:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"c6e56c186d2391d8f9cd5f65abf83b63dcc2c10b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 16:54:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"475612e34cab082204a46c9b8831a4ad1e5f9ccf","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 16:54:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ad82b381169a001a8807e6c3d80f34b12caa66ba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 18:15:42.000000000","message":"Topic add_mypy removed","accounts_in_message":[],"_revision_number":1},{"id":"d62e5b91c341422162f665f366933fa7a1b0b838","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 18:16:07.000000000","message":"Topic set to type_tooling","accounts_in_message":[],"_revision_number":1},{"id":"ff11b13079813d873ae491430038becf53dfb8a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 18:17:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17c34d8f48f7e8c8526f98985507bc21d1e90dc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 18:17:44.000000000","message":"Topic add_mypy removed","accounts_in_message":[],"_revision_number":2},{"id":"9b60a3a8103d848ff0990572b497bb53b4098817","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 18:17:52.000000000","message":"Topic set to type_tooling","accounts_in_message":[],"_revision_number":2},{"id":"414f45b894551db591c04e65b0457917e608d20e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-15 18:30:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d8aa9328cabe8bba1958957ab6db596ee0f2196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 21:06:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de9e168ce318440683754dcab37284bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7148b069d3e340c092c15a5fc280feec : SUCCESS in 3m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92b7f6ece8904cd6b13827cda41a8bcb : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4dca4f49da4a58befc98d919f8d8fa : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47e2e0fa9334f95a282f3fd80592e3d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25932a5ad10f46f2a9f1d0f2790b7771 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac58d1b6f708473fa0ee2c408dc096b2 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b08a41ae9cd4a68908330f37450723c : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99fd7dbe231f43f693b4d0357164355f : SUCCESS in 7m 07s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed5d03c331354a9580c1f350982a696e : SUCCESS in 1h 46m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe3850867b41489590e6e070878991b5 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3fe61145ae244d8b4bf25d84d68eb58 : SUCCESS in 1h 43m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8d5f81ffa0d407081a279f80b62675c : SUCCESS in 1h 49m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9d34b6096ea042babebabc62754098e1 : SUCCESS in 1h 50m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5928af6175f34521a55bf9e685a95a48 : SUCCESS in 2h 03m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42d8f24ad1be4361ad18596ed8245120 : SUCCESS in 40m 54s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a5780ac24f604681a69e555c5e4b3d7a : SUCCESS in 7m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba9aaf0dcaf248bd9197fe39b8e50c5c : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":3},{"id":"42e3e46950e2d0e90257bbe1b247812d340419a1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-16 08:31:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"badec823b98f031ca62f251611d46236af203fb2","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-05-21 02:22:55.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"b5a074cc9c9eea1223ded6baaf7c6e4ed778c5e7","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-06 21:54:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3b465dce2cd403daf9e96ee649212a70895b3c27","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-06 21:55:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ef368393c4c09af2002fcca818cd4d7777f25f5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:01:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d23f1ccfa9f0d65bb19fd242e42430e744342a4b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:27:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3635cd9b3fe32c64ed3342740b263746b986e2c8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-26 06:03:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c2eb0860cb4b4ac3fea603ed1845e6d7bc9c5641","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-30 14:54:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1220713a329eaa988c4af93d41329a55e528b1fa","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-09-11 17:15:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82d38395516447cd69d5c8c67248a849c282e80c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-12 00:34:45.000000000","message":"Abandoned\n\nI included this piece into https://review.opendev.org/c/openstack/neutron-lib/+/921484 that enables mypy for all modules.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4bd1b40073a2128216a8cd53006826440aab79b5","revisions":{"eacd746c6bfbd1e3410f2bb50321b96060dd483a":{"kind":"REWORK","_number":1,"created":"2024-05-15 16:54:33.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/57/919757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/919757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/1"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:54:32.000000000","tz":-240},"subject":"[WIP] Add initial mypy tooling for type annotations","message":"[WIP] Add initial mypy tooling for type annotations\n\nFIXME: update this commit message\nIn the last PTG meeting we have decided to try enabling\ntype checking in our codebase. Neutron-lib was chosen as\na good candidate to try this out.\n\nChange-Id: If935c4f2281938fb31e11758b7c0f6b504894833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eacd746c6bfbd1e3410f2bb50321b96060dd483a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eacd746c6bfbd1e3410f2bb50321b96060dd483a"}]},"branch":"refs/heads/master"},"b3575bf7068ccf7dae1c29363144407a7ef9966c":{"kind":"REWORK","_number":2,"created":"2024-05-15 18:17:32.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/57/919757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/919757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/2"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 18:17:25.000000000","tz":-240},"subject":"[WIP] Add initial mypy tooling for type annotations","message":"[WIP] Add initial mypy tooling for type annotations\n\nFIXME: update this commit message\nIn the last PTG meeting we have decided to try enabling\ntype checking in our codebase. Neutron-lib was chosen as\na good candidate to try this out.\n\nChange-Id: If935c4f2281938fb31e11758b7c0f6b504894833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b3575bf7068ccf7dae1c29363144407a7ef9966c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b3575bf7068ccf7dae1c29363144407a7ef9966c"}]},"branch":"refs/heads/master"},"4bd1b40073a2128216a8cd53006826440aab79b5":{"kind":"REWORK","_number":3,"created":"2024-05-15 18:30:04.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/57/919757/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/919757/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/919757/3"}}},"commit":{"parents":[{"commit":"8b6c76978995fe3b669ddf0689ac358d12174de7","subject":"Merge \"Document BGPVPN list API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6c76978995fe3b669ddf0689ac358d12174de7"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 18:28:58.000000000","tz":-240},"subject":"[WIP] Add initial mypy tooling for type annotations","message":"[WIP] Add initial mypy tooling for type annotations\n\nFIXME: update this commit message\nIn the last PTG meeting we have decided to try enabling\ntype checking in our codebase. Neutron-lib was chosen as\na good candidate to try this out.\n\nChange-Id: If935c4f2281938fb31e11758b7c0f6b504894833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
