)]}'
{"id":"openstack%2Fneutron-lib~919787","triplet_id":"openstack%2Fneutron-lib~master~I520dc744cffb542e3de84966484bcdf2aac0de41","project":"openstack/neutron-lib","branch":"master","topic":"netaddr1","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-05-29 09:38:58.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-29 09:38:34.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I520dc744cffb542e3de84966484bcdf2aac0de41","subject":"netaddr1: except AttributeError for convert_ip_to_canonical_format","status":"ABANDONED","created":"2024-05-16 06:50:28.000000000","updated":"2024-05-29 09:38:58.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"80762d0e2a0713d8a182be4d2f260e684dfb57f5","_number":919787,"virtual_id_number":919787,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-27 12:34:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 07:19:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-05-16 08:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43757c0fe9888108cab7cdeaaa85da105a7694fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-16 06:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ba03357b651074923283ba8fea7181d310b6350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 08:48:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b91cf0a53984e4dbdbf7a1972a7c841\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87de67f3eb2f480d9dc85f5726f5cffe : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b60d79605a8f4b839ab504776c56fbff : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08006760d23427c9fd1bfd2b139c342 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cc060e6cf1f4b9aac0ef909b4d0bc79 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08340a1e9b61471fb44acaf9e34764b0 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c635b3682a4444e9eceab6a04aad009 : SUCCESS in 7m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ad5bfd48ab946a8a058d62b0f9b2ea1 : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d88d825e9a414fd3bfb10954bab2b11c : SUCCESS in 1h 33m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/239c44ab0c40457a89ed665d763b9c00 : SUCCESS in 1h 51m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5878df3967c44f699fd0cfdb47a195d8 : SUCCESS in 1h 40m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/75e3db9286c94692b5b36469eba0eb24 : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d0f731d18594a22958eacecdfff81a9 : SUCCESS in 40m 22s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6d38c8a100b843579f4d558c82c3d4b6 : SUCCESS in 8m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d88d510fae7437784113ad80671524e : FAILURE in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c08d0f09eba2e5c54c3f5a3ac6f6e32672aca329","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-16 12:33:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be1d3aed47519388085b4d68c64929ff969b6bb8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-21 11:44:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"678a9b8219909a19836ab9d179a638e49ce75d8d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-21 14:08:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efc3b61db74b79334a72a280b6d6f187044b7243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 14:38:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d114f1f1d7347cf9c629473da986b7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48de0c7897b42318c4655519eb90ae6 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0c940b400844c618c8e77448f4563d5 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fe8bbb431c4d99a9c2ff8fa1379d67 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ace6492fc914b5199964bb94b513431 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10dd88a21e614d618ec76d8509786c56 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd956e586ff445f0968e53ecb1876731 : SUCCESS in 6m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a12c5cd910994645a37a26e730c3265b : SUCCESS in 1h 33m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/17ac5bda492342d4a088810af78cc0d6 : TIMED_OUT in 2h 52m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9586353bd3174b86851f9bee492682f7 : TIMED_OUT in 2h 52m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fed820eaa2ea4aaa86f18c25d73af5f5 : SUCCESS in 1h 41m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/09dd5eab302c4d34b9aadda3788c8d25 : SUCCESS in 1h 40m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f939f43dbd84459fa452c003ef42d654 : SUCCESS in 42m 51s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d4d51664de3f45aca67777647eb46e30 : SUCCESS in 6m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ea8ce40d7e44fa4a35bb1bf2e3302c9 : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"877e78a181ee1803a1143f848585f812410fa84b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-27 09:48:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42b73442829a59358371138bae3a0239e1d96364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 12:34:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f4d5b98f7347d5ae7d2f6a06855d1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb5d664e2a641198bfe2bc75ea6559a : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18e2e4fdab9c4104a38c179f64ece25b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d47c843c6d4893a067ebcc0b33223f : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d754074d471f4b67b0d495967eefd796 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/207d877f73634a3c84461cdc61a1f5f9 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/329ab0256a4f4030afe3e4fe056fc98d : SUCCESS in 5m 56s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aced05714181402ea34c736cfdb2425a : SUCCESS in 1h 30m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a3e26ad7e6184d68925de41e0db1c317 : SUCCESS in 1h 36m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2c8c3f4bc4241b0ad872418ff275dea : SUCCESS in 2h 11m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0034f44664fa4af19c8c90890376c4e5 : SUCCESS in 1h 48m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/22eb774fe4d843bfb5b12b34a97b5772 : SUCCESS in 2h 43m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a7e8bb5eff949a29d61991a2108a61b : SUCCESS in 39m 55s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a547baf173be491fa11b8f3c1641372d : SUCCESS in 5m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/170351a20594419f845154eee8ddd917 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3ca43b51e9bbcc995ac74dc46b66505c53683fde","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-28 14:04:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2a12be91d668e3575d1a0728089481e292e1337","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-29 09:38:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80762d0e2a0713d8a182be4d2f260e684dfb57f5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-29 09:38:58.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/neutron-lib/+/917263 is simpler solution.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66bdadfb4d1a257471acc55bdc216d8e8404f7fa","revisions":{"66bdadfb4d1a257471acc55bdc216d8e8404f7fa":{"kind":"REWORK","_number":1,"created":"2024-05-16 06:50:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/919787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/919787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/919787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/919787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/919787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/919787/1"}}},"commit":{"parents":[{"commit":"37e915ea1c56f2289dc911b813e317e3dd1e6dc4","subject":"tests: return same message for validate_[subnet*|route_cidr]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37e915ea1c56f2289dc911b813e317e3dd1e6dc4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-05-16 06:43:50.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-05-16 06:49:52.000000000","tz":120},"subject":"netaddr1: except AttributeError for convert_ip_to_canonical_format","message":"netaddr1: except AttributeError for convert_ip_to_canonical_format\n\nWith netaddr\u003e1.0.0 the behaviour of IPAddress changed and for\nNone as address the raised exception is not AddrFormatError but\nAttributeError.\n\nChange-Id: I520dc744cffb542e3de84966484bcdf2aac0de41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/66bdadfb4d1a257471acc55bdc216d8e8404f7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/66bdadfb4d1a257471acc55bdc216d8e8404f7fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
