)]}'
{"id":"openstack%2Fneutron-lib~921483","triplet_id":"openstack%2Fneutron-lib~master~I3c198c1695852cc7b99b08e2e49485b3a02e5bc7","project":"openstack/neutron-lib","branch":"master","topic":"type_tooling","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-04 16:50:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3c198c1695852cc7b99b08e2e49485b3a02e5bc7","subject":"DNM: mypy improvements","status":"ABANDONED","created":"2024-06-07 01:08:28.000000000","updated":"2024-08-30 14:54:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bb60105de8bacd4c3395ba7d0c01ffa3ce4b8e89","_number":921483,"virtual_id_number":921483,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-12 15:45:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-07 01:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-04 16:50:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"dd2e43014b73f36da075b9b67602755fad711821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:08:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8362fcc250e68c4a085c15861db545cb13c6a1ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 01:08:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-lib on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027check\u0027][\u0027jobs\u0027][3][\u0027vars\u0027][\u0027tox_envlist\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - lib-forward-testing-python3\n        - release-notes-jobs-python3\n        - neutron-tempest-plugin-jobs\n  ...\n\n  in \"openstack/neutron-lib/.zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"8865704c1833ef9472e0942f346265d066276efa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:13:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e08b77de299461147d2c383a059efaa446e113bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 01:14:04.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-lib on branch master.  The\nproblem was:\n\n  Configuration item is not a dictionary.  Each zuul.yaml\n  configuration file must be a list of dictionaries, for\n  example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with one\n  key (in this example, \u0027job\u0027 and \u0027project\u0027).\n  \n  The incorrect values are around:\n  \n  job:\n  - name: neutron-lib-mypy\n    parent: openstack-tox\n    vars:\n  ...","accounts_in_message":[],"_revision_number":2},{"id":"9fcfe3a3dcc316c121b0b446bf2e4df287779017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:17:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"468a2a3a30e8f011c63f914d7b4123a7601d1f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:18:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7863cd18b4670e55591dd3acc1b9e4391ac1ecfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:49:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"949a2d097ad9daaa585afb1994653273e466d9b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 04:27:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e6d2616dd9542968073d723738f1041\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/22154e65fe1a4ce1b4811312b86a18c1 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9da528309041549a0d1cd2c9fcaa5f : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3079a5b7d2a842ca930a1bd25f094664 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4036b2c14a43412c9dc63d4c48d73c87 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3320df41f76d472c98ce284c02982d1c : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/948553d6a6af4b9d868280ffffccd0fe : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9575a44ceff14ce79cf6f0a91cc9e993 : SUCCESS in 6m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8189791621344c349321a991189d9f32 : SUCCESS in 1h 49m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8da6ea6c0bb4a85a0e9267671b9bddb : SUCCESS in 4m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ad755c62e30c4a239109ffdf809a417d : FAILURE in 2h 12m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28eeaf8c0b204d4693d0c23821bc2e85 : SUCCESS in 2h 31m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3faf37c23f414d3e9f19db2e1f9ba2a4 : SUCCESS in 2h 00m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4193662c03214d308373bc4938cfbe55 : SUCCESS in 1h 58m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b03ef0ecfa43400192d6a6421d0872cb : SUCCESS in 49m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/375564edb9d945f0ba5484b1bb4dc8df : SUCCESS in 5m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fec3edff8a3641d58bdf8b70cfb94295 : SUCCESS in 55m 31s\n- neutron-lib-mypy https://zuul.opendev.org/t/openstack/build/84c04e0b5607465a825708abf0e626fc : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4038af1577f9219d26bba02039b3c4b2ac2c1767","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:50:56.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5},{"id":"8148bb3e57e50e4bb559ed95c68e1db1b8298ac1","tag":"autogenerated:gerrit:restore","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-12 13:31:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"fb38ddafbcf752153372146e9f4f4c66cc5c9384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 15:45:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03dcba7f7e2440f0870c88deee2f0d1e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dee8ed3a813b414fa0dd9e193b5a8b0d : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0964f2dfe2754f2abddb78f97ac0137d : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/196aa225671f417b8bb99ee91d4dacf9 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4054f022379e4ce0bb47d103112b5925 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8847da049b541c78b60fbd33a0a05f1 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69b92d0014524c169a82e0dbf192f3a4 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4c85c3b0aee4b2887920075157a9af3 : SUCCESS in 7m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e72e281a93248eda83049a25068e2f6 : SUCCESS in 1h 46m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/164402fa70424ba1a4b19576c277d776 : SUCCESS in 3m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4dcad1086d60495abe7b250688975db3 : SUCCESS in 1h 28m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f916a1d3dd79462991c0e09ac6c6780f : SUCCESS in 2h 07m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6a8d1a8999754a3a85059a98dad515b6 : SUCCESS in 1h 30m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ddce3c211ca4433a325ac51fcc057a8 : SUCCESS in 2h 09m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d6b75ee7fb44a67a715febe90fb016c : SUCCESS in 37m 25s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/de92227f567b4949bb32fa414b9e0c76 : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5ce5993f5898423da4cbc527c00ce440 : SUCCESS in 5m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf14f6206e304c9dbdeeaf8c9ff2c31c : SUCCESS in 53m 50s\n- neutron-lib-mypy https://zuul.opendev.org/t/openstack/build/3ef0e97dd1cf483191add21b7beb0bbc : SUCCESS in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bb60105de8bacd4c3395ba7d0c01ffa3ce4b8e89","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-30 14:54:17.000000000","message":"Abandoned\n\nI guess we don\u0027t actually care about mypy.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"67883582add6f9014e90906145c928eb34bf1d9f","revisions":{"bea9b206344c4117d38936602f48e9b2475661a7":{"kind":"REWORK","_number":1,"created":"2024-06-07 01:08:28.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/83/921483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/921483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/1"}}},"commit":{"parents":[{"commit":"4bd1b40073a2128216a8cd53006826440aab79b5","subject":"[WIP] Add initial mypy tooling for type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:06:54.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:06:54.000000000","tz":-240},"subject":"DNM: mypy improvements","message":"DNM: mypy improvements\n\n- add a job\n- switch from include to exclude list\n- don\u0027t follow imports (to avoid checking for excluded modules)\n- don\u0027t require type definitions for functions\n- add mypy to default tox target list\n\nChange-Id: I3c198c1695852cc7b99b08e2e49485b3a02e5bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bea9b206344c4117d38936602f48e9b2475661a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bea9b206344c4117d38936602f48e9b2475661a7"}]},"branch":"refs/heads/master"},"3b1d477a539b9b4d85ce2b990796b673321ff9fc":{"kind":"REWORK","_number":2,"created":"2024-06-07 01:13:40.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/83/921483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/921483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/2"}}},"commit":{"parents":[{"commit":"4bd1b40073a2128216a8cd53006826440aab79b5","subject":"[WIP] Add initial mypy tooling for type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:06:54.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:13:34.000000000","tz":-240},"subject":"DNM: mypy improvements","message":"DNM: mypy improvements\n\n- add a job\n- switch from include to exclude list\n- don\u0027t follow imports (to avoid checking for excluded modules)\n- don\u0027t require type definitions for functions\n- add mypy to default tox target list\n\nChange-Id: I3c198c1695852cc7b99b08e2e49485b3a02e5bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b1d477a539b9b4d85ce2b990796b673321ff9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b1d477a539b9b4d85ce2b990796b673321ff9fc"}]},"branch":"refs/heads/master"},"1c5d1c2ffa11698ae23c1f789b315a0d500f26c0":{"kind":"REWORK","_number":3,"created":"2024-06-07 01:17:23.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/83/921483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/921483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/3"}}},"commit":{"parents":[{"commit":"4bd1b40073a2128216a8cd53006826440aab79b5","subject":"[WIP] Add initial mypy tooling for type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:06:54.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:15:59.000000000","tz":-240},"subject":"DNM: mypy improvements","message":"DNM: mypy improvements\n\n- add a job\n- switch from include to exclude list\n- don\u0027t follow imports (to avoid checking for excluded modules)\n- don\u0027t require type definitions for functions\n- add mypy to default tox target list\n\nChange-Id: I3c198c1695852cc7b99b08e2e49485b3a02e5bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1c5d1c2ffa11698ae23c1f789b315a0d500f26c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1c5d1c2ffa11698ae23c1f789b315a0d500f26c0"}]},"branch":"refs/heads/master"},"7478cd0b954b79b918b4c183591880d8784bacd1":{"kind":"REWORK","_number":4,"created":"2024-06-07 01:18:49.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/83/921483/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/921483/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/4"}}},"commit":{"parents":[{"commit":"4bd1b40073a2128216a8cd53006826440aab79b5","subject":"[WIP] Add initial mypy tooling for type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:06:54.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:18:43.000000000","tz":-240},"subject":"DNM: mypy improvements","message":"DNM: mypy improvements\n\n- add a job\n- switch from include to exclude list\n- don\u0027t follow imports (to avoid checking for excluded modules)\n- don\u0027t require type definitions for functions\n- add mypy to default tox target list\n\nChange-Id: I3c198c1695852cc7b99b08e2e49485b3a02e5bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7478cd0b954b79b918b4c183591880d8784bacd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7478cd0b954b79b918b4c183591880d8784bacd1"}]},"branch":"refs/heads/master"},"67883582add6f9014e90906145c928eb34bf1d9f":{"kind":"REWORK","_number":5,"created":"2024-06-07 01:49:03.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/83/921483/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/921483/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/921483/5"}}},"commit":{"parents":[{"commit":"4bd1b40073a2128216a8cd53006826440aab79b5","subject":"[WIP] Add initial mypy tooling for type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bd1b40073a2128216a8cd53006826440aab79b5"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:06:54.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:48:52.000000000","tz":-240},"subject":"DNM: mypy improvements","message":"DNM: mypy improvements\n\n- add a job\n- switch from include to exclude list\n- don\u0027t follow imports (to avoid checking for excluded modules)\n- don\u0027t require type definitions for functions\n- add mypy to default tox target list\n\nChange-Id: I3c198c1695852cc7b99b08e2e49485b3a02e5bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67883582add6f9014e90906145c928eb34bf1d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67883582add6f9014e90906145c928eb34bf1d9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
