)]}'
{"id":"openstack%2Fneutron-lib~921484","triplet_id":"openstack%2Fneutron-lib~master~I285d7bfffd9349467b24d8f078847d77a5b47cbe","project":"openstack/neutron-lib","branch":"master","topic":"type_tooling","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-10 01:23:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I285d7bfffd9349467b24d8f078847d77a5b47cbe","subject":"Enable mypy for pep8 tox target","status":"MERGED","created":"2024-06-07 01:36:34.000000000","updated":"2025-01-10 01:24:58.000000000","submitted":"2025-01-10 01:23:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921484-type_tooling","meta_rev_id":"0be0fdfb830d9d8b0c56c8f75311ac663dc3a9e2","_number":921484,"virtual_id_number":921484,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 01:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2025-01-09 10:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-01-09 23:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-01-09 23:07:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 23:28:38.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2024-11-12 02:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 16:15:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-11-20 00:09:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-01-09 10:24:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-01-09 23:07:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"33eaa8e70d4eeab74ac8ceb3710dcb569c5e342a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95349602a02c055266d544acb399a54956ee1488","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 01:48:24.000000000","message":"Abandoned\n\nwill squash these in parent","accounts_in_message":[],"_revision_number":1},{"id":"f731032e8efd667a376c3c188b5c2a7277441b33","tag":"autogenerated:gerrit:restore","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-11 23:28:32.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"1a44ec9efc832cf52499bf3f8d7d6930fe8ed413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-11 23:28:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"485d6d8444c8aaca3294190e8d840c302ceda482","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-11 23:28:48.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01448724c7b2db0e6dbe9cd01a65461227ced99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-12 00:32:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a488d6a68d4dce7f9f7296a310606423a000c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 02:17:56.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40986518804a4994baae40e981187840\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d3c71f9fb37d4ffc8c5bdd19cc7c4c6f : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e166c351a9c40288592ae20ef0d488a : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e1637f2142a40f1841ac5c89769c6cb : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d61fee89b94955a44b2504eb5cf12d : FAILURE in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29f9e9f1e7f94b3d9ad26b0c5e17be42 : FAILURE in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b81ceb22c29487aac28f880a2da53d1 : FAILURE in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5b29b1326af4cb7ae509548be2e5041 : SUCCESS in 1h 38m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c0f75edf456465ab7a2a33ee80b5fd3 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/25c00c948a094cce9938a23986e75772 : SUCCESS in 1h 12m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6aaba906a074398bb5431f6e56b6a02 : SUCCESS in 1h 36m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2ca3ed1069947d791c54d00fea3f7c4 : SUCCESS in 1h 37m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7640dc37787f4e09bc609507d8e1dc11 : SUCCESS in 28m 34s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/231cfdac11af49d99c992f4767a3514f : SUCCESS in 16m 04s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/362278474982412a83ee9e82d0a6552f : FAILURE in 3m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7d08988308e423dbfc6ae2484f3b63f : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":3},{"id":"33b416fb34ebf253261f3c5951f27a2f22095934","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-03 17:45:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8234fe7bfafe4d4aac92969bb5ee25275d5436df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-03 18:18:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"51241a49523a3da8a2e21929365539483695331e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 20:30:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88804b79b25d4589b1962115d16245b6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5304aca5372f486e93197c13c73328f0 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae26d9b5b8b4dff8282aabc7eb52858 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8c2245e91fe4e4c83cf03b249acff31 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244574386b9841aab202712111be7e55 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e1e6205e21c4241b0ff78a2afb6847f : FAILURE in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96c86272712847c89d0f5162a20a7d93 : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9e039a0109f41bd815074e4ecf4907e : SUCCESS in 1h 01m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51ba011e16b54a5796d2469bfa3d7669 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b82e43009e824b068933814613f1876e : SUCCESS in 2h 10m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aee09187e8344883a7c9aee4d1dd0e15 : SUCCESS in 1h 40m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6805cbd458314ce1bc174168796b808c : SUCCESS in 1h 46m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61ccd7492d97437ab3279b3453fb627e : SUCCESS in 36m 42s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a41453c63d94fc1ba129f9d49c505d5 : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b679f661f5da4c67b2a4df5903e96bcb : SUCCESS in 2h 01m 27s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/5a5da23fab494db9a7a051433e3f803e : SUCCESS in 21m 02s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fe1bedaefb454a33b02eecf1e54bfd01 : SUCCESS in 5m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57517c7771ad4ef3b745f73ed6efac45 : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":4},{"id":"c5007390f7a37034c743ba1a0dd672b5efe726d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-07 19:03:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9d6f7ba93dd4220419f7a47a1572c7fedb4112c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 21:22:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/509e20c63e9f48d1aa224116d034896a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/99a09fc91ebf43458bd9f719ed59addb : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111c4f46e5794b0b881f41e0c20617ac : SUCCESS in 10m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cebcf541132e41b296914ef7eb0e8f55 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e00d382376481ca089fe613033fa58 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f942d65ad13b4684bef0ee76c71cf277 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b4f7b57fca64e1d8b637f40bbd11c40 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f84ba82e82149d78d39ad896b6af83f : SUCCESS in 1h 18m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a49e645b026f4192ba77072c6eb245b4 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d3d2acef70bb43fc8ad6e0438ff3aa18 : SUCCESS in 1h 28m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/46afc39049d24083ad3ebbe2df7f516b : SUCCESS in 2h 10m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3e27b8d716de45c5aa995100e2809127 : FAILURE in 1h 38m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27ef09ae12374756b1516231a0bafebc : SUCCESS in 35m 01s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b51a228ed92f4afebfb6ddd213f3e43a : SUCCESS in 1h 27m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcd23328d21443519da8246c902bc849 : SUCCESS in 2h 09m 31s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/3804a4a799514cc896611b0b7e387874 : SUCCESS in 16m 50s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/550555543b064bb884fbbf8af698f9b5 : SUCCESS in 7m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/084eab216d0649a4b1f0f8f3d957a36b : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":5},{"id":"d29ed9e216829e63e403537a58c567faf2df54b2","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-07 23:24:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c4be692ccf095cdfbc242bd494ce2534481e3beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 01:35:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693d8e02f558418a9470ab8ab2d5cbb2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f60247a43dfa49f5981126b4a5933c98 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc7aff05a5004b738c02b534f4a95434 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1391583ed3cc4fc38461741015ef1c59 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91991c13f612401e9595c1cadf7feae5 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0569ae4574f54e43a025a891f8cb236d : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bcadf04fc5745dea12c690eb8c0a6b7 : SUCCESS in 7m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04d33401ad9d4bb8942b75e44f81b13b : SUCCESS in 2h 00m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d050fcf4d0f4820b5aabaac756596cc : SUCCESS in 2m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9812d8d15310421791c61b67384c3ab8 : SUCCESS in 1h 17m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da6dc3be0e8b4c65a08b422b47f06e1e : SUCCESS in 2h 06m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37cb54b56298472d913dca1bec855615 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ea4882352f0476ab88ce53928edab03 : SUCCESS in 36m 48s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63b83e753c7b44f0821948ba21bd3cc8 : SUCCESS in 1h 28m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7647a028e044004819d1c6dc865a33f : SUCCESS in 1h 49m 58s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/7f5fee52da304bae842e0293b594c55e : SUCCESS in 18m 25s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6eb14c406f314ea3b2915b45e93edb99 : SUCCESS in 5m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/841d42e225f048358743cea9909c4b69 : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":5},{"id":"2e49ad4e4be2508cfbf45c50add57ca882938c31","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-09 10:24:49.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff7fc80e7a168c351fdffc8eeb3920e86d71e6b6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-09 23:07:31.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"013808efda7ce811f68cb84b8c2cfd9d74cdb882","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 23:08:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f9b1ed01352e3116c4824db28052db5a13f1dd20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:23:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca8f20a04b8459188815b1f315c914e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/be86e0d09562455b8b4cb62578464599 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fac596702e34afc848b30ade77f1468 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3450ab2495f94ec3a30848fddd8a34a2 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b6dd36c3759427cba20246ca34fd844 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/801eddca8c7c44e5aea5995618588c6f : SUCCESS in 3m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec76d55bff98454da983eb39774c6d2f : SUCCESS in 1h 43m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfaa652e68364375b527390f34dec0c9 : SUCCESS in 3m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8731ba6430a54b47a0042a763878cde8 : SUCCESS in 1h 28m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/459091031a3740e8b22d2264f4d4d901 : SUCCESS in 2h 05m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82b2f6eb7d574932b570e6a77cb2e6a0 : SUCCESS in 2h 13m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/364fcd3b448c47489a847988457f0738 : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":5},{"id":"1b223b0f15e09fd5569de19131d20b6c300cf45b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:23:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0be0fdfb830d9d8b0c56c8f75311ac663dc3a9e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:24:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279611e165ce49f2b967687c2ea9e20e\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1981275fa82e497d99413ef29bff498d : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19027a50c38143fa8a3111ad0b103ad5 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8130a4685904cf28f21640520b7b7ba : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3f59155effc8242eaec111f12e0653d7f8ecfdf8","revisions":{"8a4ff1a29926735e43b16ff0d18cd2a805b6e609":{"kind":"REWORK","_number":1,"created":"2024-06-07 01:36:34.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/84/921484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/921484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/1"}}},"commit":{"parents":[{"commit":"7478cd0b954b79b918b4c183591880d8784bacd1","subject":"DNM: mypy improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7478cd0b954b79b918b4c183591880d8784bacd1"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:35:18.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 01:35:18.000000000","tz":-240},"subject":"mypy: enable checks for all __init__.py files","message":"mypy: enable checks for all __init__.py files\n\nSome of the files imported from libraries with no py.typed markers.\nHence disabling import-untyped errors for the repo.\n\nChange-Id: I285d7bfffd9349467b24d8f078847d77a5b47cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8a4ff1a29926735e43b16ff0d18cd2a805b6e609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8a4ff1a29926735e43b16ff0d18cd2a805b6e609"}]},"branch":"refs/heads/master"},"2f2f67d121cfcfc240a35934809193e08d4943ea":{"kind":"REWORK","_number":2,"created":"2024-11-11 23:28:38.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/84/921484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/921484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/2"}}},"commit":{"parents":[{"commit":"c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e","subject":"Merge \"pyupgrade changes for Python3.9+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-11-11 23:27:31.000000000","tz":-300},"subject":"Enable mypy for pep8 tox target","message":"Enable mypy for pep8 tox target\n\nCo-Authored-By: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nChange-Id: I285d7bfffd9349467b24d8f078847d77a5b47cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2f2f67d121cfcfc240a35934809193e08d4943ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2f2f67d121cfcfc240a35934809193e08d4943ea"}]},"branch":"refs/heads/master"},"e7900774bb5ac141c5fe9621392be85f0e5694ee":{"kind":"REWORK","_number":3,"created":"2024-11-12 00:32:28.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/84/921484/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/921484/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/3"}}},"commit":{"parents":[{"commit":"c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e","subject":"Merge \"pyupgrade changes for Python3.9+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-11-12 00:32:12.000000000","tz":-300},"subject":"Enable mypy for pep8 tox target","message":"Enable mypy for pep8 tox target\n\nNo significant changes; put some hints where mypy got confused.\n\nCo-Authored-By: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nChange-Id: I285d7bfffd9349467b24d8f078847d77a5b47cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e7900774bb5ac141c5fe9621392be85f0e5694ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e7900774bb5ac141c5fe9621392be85f0e5694ee"}]},"branch":"refs/heads/master"},"dd72233c9cb7b76087e379d7e82e9d8ae4aa37a7":{"kind":"REWORK","_number":4,"created":"2025-01-03 18:18:43.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/84/921484/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/921484/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/4"}}},"commit":{"parents":[{"commit":"20bb2de4cf7a9a53f9a29a2c0bc91aadc94de5b0","subject":"Merge \"Support project_id and project_name in ContextBase class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/20bb2de4cf7a9a53f9a29a2c0bc91aadc94de5b0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-03 18:17:51.000000000","tz":-300},"subject":"Enable mypy for pep8 tox target","message":"Enable mypy for pep8 tox target\n\nNo significant changes; put some hints where mypy got confused.\n\nCo-Authored-By: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nChange-Id: I285d7bfffd9349467b24d8f078847d77a5b47cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd72233c9cb7b76087e379d7e82e9d8ae4aa37a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd72233c9cb7b76087e379d7e82e9d8ae4aa37a7"}]},"branch":"refs/heads/master"},"3f59155effc8242eaec111f12e0653d7f8ecfdf8":{"kind":"REWORK","_number":5,"created":"2025-01-07 19:03:13.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/84/921484/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/921484/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/921484/5"}}},"commit":{"parents":[{"commit":"20bb2de4cf7a9a53f9a29a2c0bc91aadc94de5b0","subject":"Merge \"Support project_id and project_name in ContextBase class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/20bb2de4cf7a9a53f9a29a2c0bc91aadc94de5b0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-05-15 16:49:32.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-07 19:03:09.000000000","tz":-300},"subject":"Enable mypy for pep8 tox target","message":"Enable mypy for pep8 tox target\n\nNo significant changes; put some hints where mypy got confused.\n\nCo-Authored-By: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nChange-Id: I285d7bfffd9349467b24d8f078847d77a5b47cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f59155effc8242eaec111f12e0653d7f8ecfdf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f59155effc8242eaec111f12e0653d7f8ecfdf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
