)]}'
{"id":"openstack%2Fneutron-lib~921485","triplet_id":"openstack%2Fneutron-lib~master~Icf9fcbbaa1da40194d0905829f24529cb5c65fa1","project":"openstack/neutron-lib","branch":"master","topic":"type_tooling","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-30 14:54:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icf9fcbbaa1da40194d0905829f24529cb5c65fa1","subject":"mypy: add type hints to fake_notifier module","status":"ABANDONED","created":"2024-06-07 02:11:54.000000000","updated":"2024-08-30 14:54:12.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0c9c8f3589081d42840ea4df66a830ad4ff0be3b","_number":921485,"virtual_id_number":921485,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-07 04:21:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-07 04:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-04 16:51:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"ce7da6e72e0e8da411e62c0dd4f5042d37592f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 02:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c66871923b9001afb267c7644e3f5d045867a849","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-06-07 02:16:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7064c43c3a44393b7a29c604f04784d7d68a3ecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 04:21:05.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc3de65225af4b1f9a250e5d964bbd4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c59e691fc2840de832c4154c65a4d2a : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e40af5502f84b6cb2cac6e793ea7847 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6533067f85e2474d874034e38735bc08 : FAILURE in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d447fa18641c47049e41e37f87154a1a : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc52b06991e0440b8c2857aee9445498 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b53fd2dbf8ba4349b2ab7535ad48637f : SUCCESS in 8m 37s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a387b0c5afa433e823b8c22163663ed : SUCCESS in 1h 41m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df6de2fe0222431f99d7e63d16d61cfd : SUCCESS in 1h 55m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1f88ac7c4da4eac8e9ed8fa8602116f : SUCCESS in 1h 54m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/27ad02037ebc4ec8a2f06d645a67e920 : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7fc6569ef2fd4ca78324dc285037cd2f : SUCCESS in 2h 06m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c55679021ca4ff3b20b50175401055e : SUCCESS in 38m 57s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9febbdc0c892405592bb9d887c4c087b : SUCCESS in 6m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bf041ee65934fd394e3bda091c9aa68 : SUCCESS in 52m 38s\n- neutron-lib-mypy https://zuul.opendev.org/t/openstack/build/5e37d68bc30b45c685239ea69673185e : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fc7eb0e4d411df3e294a94bb7a7999ed6bc93fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:51:05.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"a03ef64e891a05d90633cb1f5ddd8afa066dc901","tag":"autogenerated:gerrit:restore","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-12 13:31:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"21ffd8e426317c33a0d624c7bf58626214834ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 15:30:52.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db7cb03086294f9d818c02dd6b151dbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77332d27717a4e9eaa21b928bed31278 : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce73f59de7744a5a9e908973dfc009e : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc1d73cf398475c83cb2ff94f3d0b87 : FAILURE in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/725670bcec044cbbb71c5e7e34ec6a10 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7749e046bd854a938a96f89e42f1acc8 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16fec89c96394e8d843bff6ca320201b : SUCCESS in 7m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30604424a4c142db901bf4a511a9596c : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/40f927009e71491a9864b058b31e180f : SUCCESS in 1h 51m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a8a26732a63487e84f85254ba80a092 : SUCCESS in 1h 55m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8f59cf7c00e04292b41d0e41ba0210df : SUCCESS in 1h 33m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0dd9ce20c2384caba93d00a9ae4d5653 : SUCCESS in 1h 42m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa2a47f35548448eadf3d9b1ea8701ad : SUCCESS in 48m 32s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/d69a2311a6464276843735fed59e397c : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f4c6ca0f8cbd40b4895747c2d1aacb81 : SUCCESS in 5m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13af80a7d60a4334817a771cc3d30956 : FAILURE in 53m 26s\n- neutron-lib-mypy https://zuul.opendev.org/t/openstack/build/f8d4af7bc97e4c609a7d2dc381752aef : SUCCESS in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c9c8f3589081d42840ea4df66a830ad4ff0be3b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-30 14:54:12.000000000","message":"Abandoned\n\nI guess we don\u0027t actually care about mypy.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19206276c6a21fa1a3b8294804da2d5afd6c591f","revisions":{"19206276c6a21fa1a3b8294804da2d5afd6c591f":{"kind":"REWORK","_number":1,"created":"2024-06-07 02:11:54.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/85/921485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/85/921485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/921485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/921485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/921485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/85/921485/1"}}},"commit":{"parents":[{"commit":"67883582add6f9014e90906145c928eb34bf1d9f","subject":"DNM: mypy improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67883582add6f9014e90906145c928eb34bf1d9f"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 02:08:56.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-06-07 02:10:37.000000000","tz":-240},"subject":"mypy: add type hints to fake_notifier module","message":"mypy: add type hints to fake_notifier module\n\nChange-Id: Icf9fcbbaa1da40194d0905829f24529cb5c65fa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/19206276c6a21fa1a3b8294804da2d5afd6c591f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/19206276c6a21fa1a3b8294804da2d5afd6c591f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
