)]}'
{"id":"openstack%2Fneutron-lib~922085","triplet_id":"openstack%2Fneutron-lib~master~I744343804cd6b6a4bc7353502d7a0295800b37bd","project":"openstack/neutron-lib","branch":"master","topic":"bug/2069595","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-06-18 20:09:23.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-06-18 08:14:08.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-06-18 20:09:23.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-06-18 20:09:23.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-06-18 18:07:02.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I744343804cd6b6a4bc7353502d7a0295800b37bd","subject":"Add the ability to define the ``BaseWorker.set_proctitle`` flag","status":"MERGED","created":"2024-06-17 10:41:06.000000000","updated":"2024-06-18 20:10:21.000000000","submitted":"2024-06-18 20:09:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922085-bug/2069595","meta_rev_id":"0eb4c8254fd868468280a3a4287d9faa19e1d7f8","_number":922085,"virtual_id_number":922085,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-18 20:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2024-06-18 08:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2024-06-18 18:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2024-06-18 18:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 13:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-18 06:18:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2024-06-18 06:18:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-06-18 06:18:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-06-18 06:18:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"dd24db10d5ffd5b7f9323085aeb482210bd4fb01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-17 10:41:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae72aa7122ec9f02cb90c56c6c28ef0e9ca0f94c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-17 11:47:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e92db5eecfc16101794b399354fd51aeb90d0d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 13:53:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c8f8c3f907a4e88a0d4601cce8b5461\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d24bc85cc904101b5be776a595f9e63 : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb79d644562484ca42a914a8c2bde88 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc49b69e4e34c78995484115c212808 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ed001706cd746528f795b5010b80b8b : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d738af5ffd4c4c9ba7c122b8ee9c23ad : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3473b9f0aed478d9fb73e72b743368f : SUCCESS in 7m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4fcc4c154d84e0d892179a04828b0ae : SUCCESS in 1h 45m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8ea5118e9b6f4f87874db34e3616ef53 : SUCCESS in 1h 37m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8184a3591fc644f9a5e4753af60a50f1 : SUCCESS in 1h 44m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f11b59b2128b4bc8bf76f51bb017f074 : SUCCESS in 1h 25m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/be659e25cfca4789ad55fa104aceab30 : SUCCESS in 2h 04m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77be9e78e4774f1d9db6f896408a6172 : SUCCESS in 33m 30s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/6fb79c6abf804a01b8be9f2e689493fb : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7c37229fbce04697a409c6e6f87e89ce : SUCCESS in 5m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/217c73314078456eab9d74b35fc7a544 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b016ace95b7958296d56d65529a6104584effb71","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-06-18 08:14:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"69133075f6877014adc9c8aaa10932e9fc2a060d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-18 18:07:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e77b28fe26926adf226ea8e8a81d91ae62353c22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 18:07:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc86dedcb6f5ebbed664261fdc2db31369d9832f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 20:09:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8ccdc4366a4e0583c157d727403a57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c8cc20f8c14cd285990dcc629f0d21 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c91e432dc294aa48a70d8b54a22ed6d : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca381d4371934887afd3fffd6b3bb560 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e01971cbd50c457eabcb0585aa14d399 : SUCCESS in 4m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78a6b6311f794599bc8c380261d8befe : SUCCESS in 1h 48m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a70529834d254855b817341fe134ed7f : SUCCESS in 1h 36m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52b19eb7fa7c44b9a1b3cba8c8f59304 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc54dff15c50499fbd09334988987b12 : SUCCESS in 2h 01m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9c883d2b914c4704ac454ecacd0bda23 : SUCCESS in 1h 51m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc95c0dcea85409f9b28f2cac30b1eab : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"30e8c50bcae16f4813050b9d4c3ced4d4fbcc5bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 20:09:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0eb4c8254fd868468280a3a4287d9faa19e1d7f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 20:10:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cbd45f1e1f5465d8f3bfa42ef3b8560\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89933422daef4edaa136d71c5b9c9340 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a13651df89ae9b543e7bbd03f4f4ad0834dc08d","revisions":{"76f90cdd1951ed194adcc93943995f0f7644c424":{"kind":"REWORK","_number":1,"created":"2024-06-17 10:41:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/922085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/85/922085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/1"}}},"commit":{"parents":[{"commit":"7674b4b6666df1c1dac4ba28d88a34f389138b52","subject":"Merge \"[api-ref] Add note about no validation of the target_tenant in RBAC API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7674b4b6666df1c1dac4ba28d88a34f389138b52"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-17 10:35:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-17 10:35:59.000000000","tz":0},"subject":"Add the ability to start a ``BaseWorker`` defining the proctitle","message":"Add the ability to start a ``BaseWorker`` defining the proctitle\n\nThis patch adds a new variable to the ``BaseWorker.start`` method\nto be able to control if it is needed to define the process title\nor not. When a ``BaseWorker`` is spawned in a thread instead of a\nsingle process, this worker should not modify the process title\nthat should have been defined by the main thread.\n\nThat is happening, for example, with the workers spawned in the\n``AllServicesNeutronWorker`` class. This class is defining the\nprocess title as \"neutron-server: services worker\", but the new\nworkers spawned (``PeriodicWorker``) are re-writting the process\nname to \"neutron-server: periodic worker\"\n\nChange-Id: I744343804cd6b6a4bc7353502d7a0295800b37bd\nRelated-Bug: #2069595\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76f90cdd1951ed194adcc93943995f0f7644c424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76f90cdd1951ed194adcc93943995f0f7644c424"}]},"branch":"refs/heads/master"},"2a13651df89ae9b543e7bbd03f4f4ad0834dc08d":{"kind":"REWORK","_number":2,"created":"2024-06-17 11:47:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/922085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/85/922085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/85/922085/2"}}},"commit":{"parents":[{"commit":"7674b4b6666df1c1dac4ba28d88a34f389138b52","subject":"Merge \"[api-ref] Add note about no validation of the target_tenant in RBAC API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7674b4b6666df1c1dac4ba28d88a34f389138b52"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-17 10:35:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-17 11:45:34.000000000","tz":0},"subject":"Add the ability to define the ``BaseWorker.set_proctitle`` flag","message":"Add the ability to define the ``BaseWorker.set_proctitle`` flag\n\nThis patch allows to define the `BaseWorker.set_proctitle`` after\nthe initialization, needed to define the process title or not when\nthe process is started. When a ``BaseWorker`` is spawned in a\nthread instead of a single process, this worker should not modify\nthe process title that should have been defined by the main thread.\n\nThat is happening, for example, with the workers spawned in the\n``AllServicesNeutronWorker`` class. This class is defining the\nprocess title as \"neutron-server: services worker\", but the new\nworkers spawned (``PeriodicWorker``) are re-writting the process\nname to \"neutron-server: periodic worker\"\n\nChange-Id: I744343804cd6b6a4bc7353502d7a0295800b37bd\nRelated-Bug: #2069595\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2a13651df89ae9b543e7bbd03f4f4ad0834dc08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2a13651df89ae9b543e7bbd03f4f4ad0834dc08d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
