)]}'
{"id":"openstack%2Fneutron-lib~923926","triplet_id":"openstack%2Fneutron-lib~master~I17a455e19c4bbbc45e26dee88849ab5e0b4b5d65","project":"openstack/neutron-lib","branch":"master","topic":"bug/2072567","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-23 17:55:44.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-07-23 17:55:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17a455e19c4bbbc45e26dee88849ab5e0b4b5d65","subject":"Change the ``StandardAttribute`` load method to \"selectin\"","status":"MERGED","created":"2024-07-11 06:32:49.000000000","updated":"2025-02-04 09:49:14.000000000","submitted":"2024-07-23 17:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923926-bug/2072567","meta_rev_id":"76cdf43b71bbe033fc9def7cf0538b410e548636","_number":923926,"virtual_id_number":923926,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 17:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-07-23 07:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-07-23 15:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2024-07-18 10:17:12.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"date":"2024-07-22 15:02:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-07-23 15:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 08:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-15 13:17:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-07-17 08:27:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-07-17 08:27:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-07-17 18:47:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-18 08:40:28.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-07-23 07:27:48.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-07-23 15:48:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a82c25171384b04b0427378c64b94d62cb2f1cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-11 06:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7ee15a17f98d44507e479eed05f7a2690bb0ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 08:48:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27ae4f352a744b6ea175a65c4387a978\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca9743e0a4c9416d8ecaee79ec9f3c92 : SUCCESS in 4m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1d9c3c0bc4a41eca417dd68c53724bb : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b698a6e7a64f5f90563c2ae918c021 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/742380244cf448de820477cc016faf4d : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9939191255e9456d8aebcf8cd33b155e : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49d1bee2145c404cafcf478a98350845 : SUCCESS in 7m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/457a6a0780004ba3bdcdff9cfcd9e8cc : SUCCESS in 1h 35m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e0b006053ab4c2ebd4a489a21511455 : SUCCESS in 1h 33m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1a8ff8cf7dc4ae58eff5298d56e2746 : SUCCESS in 2h 14m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b110e5c876cf4e23b0ea5bbb19500e88 : FAILURE in 1h 49m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a55798041098446b87df65270e75b340 : SUCCESS in 2h 05m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/85d38ccf5e564f2abc0b273897ff8dff : SUCCESS in 41m 26s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/b7f4a6e06cec465f8bb8186079ec73b1 : SUCCESS in 18m 27s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8ab5abced94e40edb93a0103530b28f0 : SUCCESS in 5m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11f1575e6d91495fb6a25d80591e9a3e : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4c519a1859f90db3b153855f011104f58f211420","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-11 11:28:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c9f501060dcbaa200ccd88f26d3d8487182ba6a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-11 12:51:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03153b664d073a8b1c198cb7475749d47a21bce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 13:53:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4848535de476426dba3e21a6e2ff766e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87711e6f5b34d2891cc5f4e5e1b37da : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e493689a237141ef9246ee1812eb5aa7 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b804751a4ecc4a86a8d9303aee1c4ff8 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4f3f5a59294b48b943e51ed2d814df : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a59b460ad7454ffc8a40e6deeb946230 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a40f684d21fa4ccb969b19820d8b0a61 : SUCCESS in 6m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e8472b21d624d97a6ef78d90c4e5c20 : SUCCESS in 1h 37m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/590b23166b85482388774679c433b762 : FAILURE in 1h 40m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d7ef235bdaa42699b6b7425c7f5c817 : SUCCESS in 2h 07m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/37ef6f25d6884766bd18fb532849b0da : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4cb54242c3e9463a848b7b1035d19aa8 : SUCCESS in 2h 23m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27afba043b0b48cdad2d0871966e990f : SUCCESS in 35m 27s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/dbbbe0d2fd134775a1137256bc67a3eb : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2143fd44adbf436c8bf6b6705b3cb3ca : SUCCESS in 5m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ffa225824dc4c1ebe39ab22c502ee84 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3f0c26dac12d2e442774a3975613c343bf4fa9e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-12 08:55:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14656b75a726fe76fb0006ff493eb9eeda414dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 11:11:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd66c4021eaf442c9279311cf7094de4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cdc836274444c0f9ab79f55054d2a74 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2507c7c458fb4b9089be77f1912b48e4 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417e68758a7d4d50ad93669657aa2944 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc0cec9c16a4b06a9b0b154ac2c4394 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb180020a1034dddb5527ce392e6782d : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52a44564420d4f1482412625b1b1226b : SUCCESS in 8m 34s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e3fd1a9028149ad92f999f01b178aa2 : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7787414809af4226a6ac822f022f9efa : FAILURE in 1h 35m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7f6fe0a286ce4566a9a16eae9ca3a36a : SUCCESS in 1h 39m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0f3e44995c72489a9c4e7462454e6ab4 : FAILURE in 1h 36m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c065a511f344c8290a7b1c73e722eed : SUCCESS in 2h 14m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d896bbe896884bff80d8caa5a18d9727 : FAILURE in 49m 18s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/057ae9da8410454ba310e84e4518ebf1 : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/caf984faccf6483b91dd33494246acb5 : SUCCESS in 5m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a78618078344f128207331676f87511 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d957695af5b3f9e08907f433c7b2a67e60512711","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-12 14:23:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9104512e474befe12935467451b1db1fcbfbbfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:51:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/265e5be24a004721afc17b28f691ac5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80f8c58bfe94bf394661f3a65e5ed96 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e02f88b399cf4b05a978fee42257b944 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15db12e045b84517bd93f193dbefc7d8 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/610f531d2a494ca0a8bfc4352e082c63 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d308f575e6564ecc861542d52ceda723 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80238273361a400ea9cc55bbc1fe1d23 : SUCCESS in 7m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f7b24582e73499a94644300bd1090aa : SUCCESS in 1h 33m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4b14d637ed2b40338383ab935356688f : FAILURE in 1h 49m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a85e5e11ca44d8da00c7847242afacf : FAILURE in 1h 49m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/62839a77850747039405013bd06b3467 : SUCCESS in 1h 52m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53cb980a571140008b7977abd3412e40 : SUCCESS in 2h 10m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b7444ae60f641238b91d443d9000d3a : FAILURE in 1h 00m 05s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/f113b66dc91b4a128ab03f1adc7cce2d : SUCCESS in 20m 13s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f023f92a48174d56990097cf6f749d12 : SUCCESS in 5m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4f00eea47ce45a1a80d4a39fdbb14ad : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d1c537c264543af09aef4ed19da45360023d6d4d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-15 06:48:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a052f584139d8817057e33437e9e10f42402c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 08:52:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79e908062c6b40f6852d3a4832765d48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28361619f33d47ca830059b435b09495 : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10a60fe7f4644f539abad338efed1332 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ee75d4833b4719939b93a6b48f7327 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39bfbbe78a9348b0b073075e7371fa9c : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcf0c1ec41e24c3caf4ca36faa74a807 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4c692799e5743e98ebc0f7096e32451 : SUCCESS in 7m 24s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9bc017f16464367a78f72dee1199a81 : SUCCESS in 1h 36m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4d157e7e987e45ce83c132756c9670b6 : FAILURE in 1h 33m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c0ab1c09e9014121ad0a34f7c28c7ae1 : FAILURE in 1h 56m 42s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3d477cdd6f54419b8d0e057f56ebe80d : FAILURE in 1h 38m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db2ac2411c744aa8903773e0698848fa : SUCCESS in 1h 37m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf606d4b412b4f3d99d25367c6bc12ee : SUCCESS in 39m 37s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/6745a126dcce48669a67691d65250a66 : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0c50be490c884dce9b5fb58e7a1efe3c : SUCCESS in 4m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d6c84d34adb4e0a86236c559f8b76c9 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0f878adec45ff5aa7d8399e829800309eb34e15a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-17 06:50:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01b0d885302ea6d1f5338f831105aa1f09af2070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 08:51:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd69a43312324a338079b8700790f76f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8566200e2cc14bdd81125e5fc4155900 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ceb9d233a974dbcbfd130a15ddafcbb : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85a9fe476f74435ad05dc03b5163e1e : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/752fa2af53664ac8a01672ad258d7a6c : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f150edd582864fe6b19a76ef906bf8ac : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d77cc1f4b5614d7dae4fb2bd44fb17ad : SUCCESS in 5m 35s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/518b44ac8f8a45678ae6377238f30783 : SUCCESS in 1h 54m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b45efa24dc0b48fdb6107f28168785b5 : SUCCESS in 1h 47m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b0f237e9455445eabc9bb6d2cfe6e36 : SUCCESS in 1h 43m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1208fce28288429aaebc7467cb3547ac : SUCCESS in 1h 21m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8c2daf0cb67247d996221facc8412e7b : SUCCESS in 1h 55m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4cf5492e1bb64714a09927b801c3c8a2 : SUCCESS in 40m 10s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/bc383dcf1c004edc9e7f7d2189434d1e : SUCCESS in 19m 42s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/54dcb1ba81ab419abc29bd684f9eb336 : SUCCESS in 5m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00c1ccc73f7a487182c533d3b5158455 : FAILURE in 37m 15s","accounts_in_message":[],"_revision_number":1},{"id":"61dd029b06b9e3a504a015a46f1ee0a1b5bd7f92","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-17 10:16:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d364a4a55247667e60c94986e70502bf0763de86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 12:16:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5426ff4a2bfc4d6d95e890cf36a735b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8e8ad83ec249ad93ccdf5afe9e98a1 : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e86dd426dec54b809035eeddddefe4fe : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52000174f6af4f468d057c5d7f07a547 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adbfca443c9d4f35a56e1b27f9ad0b0d : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/854f26fbbd6a40869b6f6c34b33ace3d : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/609cb5ccc892436690d63620ae46c4ca : SUCCESS in 8m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f9109a9fb494c3ba51c0d33afbbd110 : SUCCESS in 1h 32m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/192eae62426742b6b8753cbad09a5efc : SUCCESS in 1h 26m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f261a6da6f4b402bbab5d65c86c6b545 : SUCCESS in 1h 57m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8597c6b7f9ef466cb572162b89cddf92 : SUCCESS in 1h 55m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd8086b35b164b2b88e9f3bc60e9599f : SUCCESS in 1h 37m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b86f61b52b3844228bfffa3bfe96a41b : SUCCESS in 32m 29s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/80b0e60af181481489c53d24c6bf7ee7 : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2d2684fab40d4f2a866235846266883d : SUCCESS in 5m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd1f4b82745145c6b94e3fd3657815bf : FAILURE in 50m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6025096fc535f90d283c71505dc541ff3b27b6bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-17 13:04:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4244748cf38d7135585f63982a86efca90722d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 15:28:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c044d9217694605a557b41d85caf3f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d190fa94b2745378dacc69bd813f73f : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e89f5f4a6a441efae2004aae74b16b4 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af90e96c6b064aacbd74eb51027d7c4a : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db7759a8a5c444108c5d3ad8ac16c8c7 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/797a6204914f435f917b06737db1d6dc : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/150bd96c3b9b4d6195ae264246a2f516 : SUCCESS in 8m 15s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e4b580e2817466caf041385b687a2e2 : SUCCESS in 1h 30m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8bca2ea524d64345b8d3e42486b0eaf1 : SUCCESS in 2h 00m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1dcb7cbfc85645f2a63b78a8452f6fa5 : SUCCESS in 2h 08m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a1f947524fed4ce4bc803f76516bcbc3 : SUCCESS in 2h 21m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/247a2c9698fe4109b543fc226db9aed2 : SUCCESS in 2h 16m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c8adc6d91a847afb6210789f63e5fa2 : SUCCESS in 45m 54s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/986ac741b22e47f6a719fe7562f4a215 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2c3c79d65fb94338ab6cb2f8ba4e8dff : SUCCESS in 5m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f2d3332adb84a309c1f34e249aca4c2 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"35f8a7683872c7f162bc45268e9c8cfcc834d859","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-17 18:47:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ac4f0bb4ea99195beb056df4456a2549a9fcf9b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-18 08:40:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0876714830504a78cadb79aefc91330da1164c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-18 10:17:04.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e938029d3343688aade9b3eab5267d023b6146d3","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-18 10:17:12.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_1131\u003e\n","accounts_in_message":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"_revision_number":1},{"id":"6a31d58ce17274f4e6dabc15f969f487b2f71946","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-22 15:02:03.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4de64057322c6fafbf5ade51193b95ccddf3c68b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-23 07:27:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a73285230669a269feff80f05556a76ee7252dbc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-23 15:48:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31d0243c943026f03ce840496fd053daf8408643","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-23 15:49:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8a53094a704f9d5a577b151d1ddad75458163eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:49:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b57873acff5172e4b0d6dca0e63bc99804b6f6d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:55:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c17951a4aaa741c8aed57b1dc252a842\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cb0a1e15594523a6b97439dc4c18ee : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1726d4905747399cf2cffae298c2e1 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e3f02b5ef7f4f0ba2bcaa61268a8181 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92ab3920a28b4fbe892ba51d1dd38d74 : SUCCESS in 5m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e8ad4fc16ba403ea0848c0dcb17aff1 : SUCCESS in 1h 33m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b35c4fd3820147a39ba301479151f02f : SUCCESS in 1h 20m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b7ec8c1bee16437bb54e18dcc6c1e509 : SUCCESS in 2h 05m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26f245b77d3a40c58931856dae430b42 : SUCCESS in 1h 55m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/75777bab0b964a2883a77b3ac33c3a55 : SUCCESS in 1h 45m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b749725d274445efae1ba4f6242ed218 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"381735ad65806d2f7eff14982814afa18053fc24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:55:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf2fc0e5518ec49991d2600d1b3678fabc284c7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:56:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce1299fb70e4eb9add70f98392e7264\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c85fd2db15c468d826c020f63c33195 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"76cdf43b71bbe033fc9def7cf0538b410e548636","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-04 09:49:14.000000000","message":"Created a revert of this change as Ida5afffb808597aa09cc62984b36189a8bf409a0","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4","revisions":{"7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4":{"kind":"REWORK","_number":1,"created":"2024-07-11 06:32:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/26/923926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/923926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/923926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/923926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/923926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/923926/1"}}},"commit":{"parents":[{"commit":"88ecb3cf96564b202b92cb8d819b4e675b390cf2","subject":"Merge \"Add start_rpc_listeners() to MechanismDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/88ecb3cf96564b202b92cb8d819b4e675b390cf2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-11 05:58:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-11 06:31:26.000000000","tz":0},"subject":"Change the ``StandardAttribute`` load method to \"selectin\"","message":"Change the ``StandardAttribute`` load method to \"selectin\"\n\nThis load method will remove the ``StandardAttribute`` load from the\nresource main query and PostgreSQL won\u0027t require it to be in the\nGROUP BY clause.\n\nCloses-Bug: #2072567\n\nChange-Id: I17a455e19c4bbbc45e26dee88849ab5e0b4b5d65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
