)]}'
{"id":"openstack%2Fneutron-lib~925503","triplet_id":"openstack%2Fneutron-lib~master~Ibb919c53d7e1bc2710e85ae86a0ed0d7620a900c","project":"openstack/neutron-lib","branch":"master","topic":"fix-internals-docs","attention_set":{},"removed_from_attention_set":{"37208":{"account":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"last_update":"2024-08-20 16:11:16.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-08-20 16:11:16.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-15 17:17:43.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ibb919c53d7e1bc2710e85ae86a0ed0d7620a900c","subject":"Fix python modules docs build","status":"MERGED","created":"2024-08-01 15:34:30.000000000","updated":"2024-08-20 16:12:35.000000000","submitted":"2024-08-20 16:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925503-fix-internals-docs","meta_rev_id":"24ca11dac0af08b20c6d69066a12de7f0050a0e8","_number":925503,"virtual_id_number":925503,"owner":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-20 16:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-20 14:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-08-15 17:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-08-20 14:04:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 17:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 17:09:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-08-05 10:59:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-08-09 07:05:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-08-14 13:22:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-20 14:04:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"6e2c7d5b832b9aafa29d52a536593698b2aa5c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-01 15:34:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a64b277211a28cc8020e250fd7bf41c2c0bd8ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 17:07:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/805ad60dfb9b46adb7530a16b5f749ed\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2897bc9827ea4e3ea3a90d9485301432 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eedbe4797768418d9e949aabca12e1ed : FAILURE in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bf803f7f7704870ac5547262445bc41 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15aca23015724a4d952ed41277e3bca0 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52020e4206584ee2be2b6f6d93b583d4 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6b2fa1c1e3d4238adea404f8dddda0f : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d21d8809d83f4256aca8fa51bb6d5ac5 : SUCCESS in 6m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/640559cac0ce421cae9a06494deea9bf : SUCCESS in 1h 30m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05dd5378f23d43d18e0d2ee530c0537a : SUCCESS in 4m 50s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/a1cf48079f584846979a66954dd76322 : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2ee3f62ca67e405ca4a75abc15380f23 : SUCCESS in 5m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd578928dbe24b299388d8cb93d8102e : SUCCESS in 45m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8de84ffe6661d7abdb139b61f0f7df04827efafe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-01 17:09:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0f085a90a873d1fe5a7d319905b1edf337573cb","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-01 17:45:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8232b74d17a477fcf8abf039155d50e759bba0f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-01 20:29:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"164ac0473d6834abf10ec1cf53238dc3e122c3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 21:47:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee22150e76e4db4833e467fdde95553\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5a2725f1e3d0480f941fc9869e241b23 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3f273ac077446c98e2b83910cdbb2e : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c33260983ac47aca848668cf5cad6cc : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f3109856cf4e479a89b6e4a17421ad : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d872448d674948bb8639d0d5cccb0407 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6778eb686d4d4080b532a0a7fc45b0e4 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3526c89f25914b2180e22b74e2dc2974 : SUCCESS in 6m 46s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e4d844b69e948e2bf55db2901578ef8 : SUCCESS in 1h 16m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eafe35737bc4087813079cda5e6f8a5 : SUCCESS in 3m 31s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/9a174fc248a6498e8badb26bd7e1d0ca : SUCCESS in 25m 40s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/10279fb308d74020afd2a720d1525644 : SUCCESS in 5m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8aa96eebfcb9474dbcf6c08f59b1b01c : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b446829e057066f16be2ddb6925231749cb061ec","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-05 10:59:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5f5e5667ca2cc3a5e1374e693f8b824ffe44fb4","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-07 14:05:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d7020457e8d29e3b0924d93b78147d653edc863","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-09 07:05:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bdaa7c4b6742b66019f41b840065fbc83a99992","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-14 13:22:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c11e5c8f1f04201bd8816f91315c1a64ee27d8d","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-14 13:41:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"851c15c4bcaddafa35cd80a378d90b07a450104e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-14 13:48:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e23fa9c275a9e38294d8db99f814958bdf523f2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-14 15:44:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5972ee174b290228d03f2825ce27e68e40a2cc74","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-14 15:47:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd240105d412095dc0b4f72f30de83ae0ef51246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 17:31:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b439b95da341e0bd2ed26360558b9b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5eccf03ee2ef43bd89030c78af48df81 : SUCCESS in 4m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/331dead4963d4213b0880f365ef91561 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc471877184843558ff36b744558d498 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fb70eaa7dfa49f39f1c1594dc999510 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1fbfde6fdc48fea9e66621ade89ff4 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ef1fef0b0574e4b9e028bc49e96f80a : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ab0e5411ba04a658e7309bd862b588c : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/147db57242fb45ef9b3b786a3920ea9d : SUCCESS in 6m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca91f9647fc1418ab9034b6aab831d36 : SUCCESS in 1h 41m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23a457183fa9420498f4379f0c7271c8 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72b31a08912e48d58690e104cc382799 : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac19c4b3531946d6b35a9175256912d5 : SUCCESS in 1h 35m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1322024dd9d84f13ae985c899e54d9c9 : SUCCESS in 1h 24m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6f2747d376a45749dac16f2329ad223 : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af314dd56b724937a10cd9eef9d9d07d : SUCCESS in 37m 17s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/2bd9877932ac498a83168841425b406e : SUCCESS in 25m 38s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5bd91a01aad3414da2babe4a04b9d124 : SUCCESS in 5m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b7e3cc29dda4a46b6543a7cb4f0dbe2 : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":3},{"id":"efa4e16416709e48f7136428ed62012e6bc54917","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-14 18:13:34.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b4ffdd403a5d1d35887f5f2d67433ef8bec4c9d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-15 13:28:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9b85079d23bdec92a4eb34b7da48eab7dc7bb96f","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-15 13:28:11.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4777b9b5e6dded0987f068642e32125d1a1a6ef9","author":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"date":"2024-08-15 13:28:25.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"410df0f6088f85b3d76931a59157283660cd8879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 15:46:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a6921a0676c45b3be8eddb282ef1999\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03a8014e98d94e39828352c1092d6506 : SUCCESS in 3m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02ba24db44a64fdaa082ded4dc7d078a : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5831dae2bbb74ebba03298e0a1758afe : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b19fbc8a9134013bc7d933bb7097d85 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06e6d1e92044b019b475b8f1522ef62 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/473fe27990b34d65a594aab15362b29e : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12b6e35c2b66418da271bf74229b0fa8 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf2e713a96514f83b8f841e34842783f : SUCCESS in 3m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14303c23d9d8489087e5a4649b0d2e43 : SUCCESS in 1h 35m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19170a391c464c98bd8005a5fe8b8874 : SUCCESS in 3m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/873ac5dae2c044a2bd96190c0d6cac88 : SUCCESS in 2h 13m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7792388ef69d448aab69498a704c40f9 : SUCCESS in 2h 06m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/144e8ee19e824c59811c72a7d45d384d : SUCCESS in 1h 52m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/09bb74a566bf425c90e2bb920b0c5d44 : SUCCESS in 2h 16m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef18c8d3d2a8402ea32b4f6a0869ecd7 : SUCCESS in 39m 47s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/8d606d7d8e2346e9895609bd6793e882 : SUCCESS in 19m 09s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ecb9f358a9fd4781b6fa37f559cfc40d : SUCCESS in 5m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac23ebc4f0d241e78bf2fc532f041ac7 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":4},{"id":"abf3955c1bbb32a0684f9745f8f124bf2b87f3cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-15 17:17:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1a90913874f32f32fe8e62958b9113f47b757740","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-20 14:04:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"15575be8c23b5356a6222101c235e71508eb72d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:05:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6bc1e67db12931be324c1d5bea7a9c5c33df8a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:11:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a883c446e9404994bfb5238a9a0ffec6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1f33addcf064a3e859a83de1d334ad1 : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/be69ae52809049679f5354713acffc08 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362bceb706f84b518cfae10783dbb2b8 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e484b6a334b42048ee0afaf406530d6 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e304cc3514c841889215cfb478d31c67 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a6801cfe88442ecb434f9e5036ec27e : SUCCESS in 4m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0af1e8bee64841deb597a851a02b9e8e : SUCCESS in 1h 42m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ed10c69b974ed9845bb21a4bab5d7f : SUCCESS in 4m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b8cc10d9277447a8ce6a435cbbd87e1 : SUCCESS in 2h 02m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1b0221f7bbe4149b73bd63a99b488c6 : SUCCESS in 2h 04m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/77609ca2b79b4f5db3dab8f11f02dd69 : SUCCESS in 1h 33m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/892dca14aa604985a2f989d93075977d : SUCCESS in 1h 49m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfc3fc952c894a45b63e9d6afa2e43a9 : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":4},{"id":"bec37ca973d25b69e0a78b8ad97dd30199d8aad8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:11:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"24ca11dac0af08b20c6d69066a12de7f0050a0e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:12:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbbbf4ba93b849fca7dc3e8a3e6097cc\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7336f6bdda5b4cb0b188bbd23a2b3bca : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f87557f9fee4344a05f15f1442e40aa : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f156724cae54eae9bf958147c227d1e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d69eae326e380440fb7acdf7c7a233eed43c3f32","revisions":{"0cf152a00a57ace5227d62b7a50694aee76ea86e":{"kind":"REWORK","_number":1,"created":"2024-08-01 15:34:30.000000000","uploader":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"ref":"refs/changes/03/925503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/925503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/1"}}},"commit":{"parents":[{"commit":"90dcbe1f6be04aba5ac90201045be3721c06625e","subject":"Merge \"Change the ``StandardAttribute`` load method to \"selectin\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/90dcbe1f6be04aba5ac90201045be3721c06625e"}]}],"author":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-01 15:34:15.000000000","tz":60},"committer":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-01 15:34:15.000000000","tz":60},"subject":"Fix python modules docs build","message":"Fix python modules docs build\n\nCurrently the Python code documentation is not being published at all\nwhich can be observed by visiting\n\nhttps://docs.openstack.org/neutron-lib/latest/genindex.html\n\nThis used to work in the old versions, for example in 1.9.1:\n\nhttps: //docs.openstack.org/neutron-lib/1.9.1/genindex.html\nChange-Id: Ibb919c53d7e1bc2710e85ae86a0ed0d7620a900c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0cf152a00a57ace5227d62b7a50694aee76ea86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0cf152a00a57ace5227d62b7a50694aee76ea86e"}]},"branch":"refs/heads/master"},"8cf30e80859702c317447bbb95f22ed7bacb1105":{"kind":"REWORK","_number":2,"created":"2024-08-01 20:29:44.000000000","uploader":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"ref":"refs/changes/03/925503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/925503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/2"}}},"commit":{"parents":[{"commit":"90dcbe1f6be04aba5ac90201045be3721c06625e","subject":"Merge \"Change the ``StandardAttribute`` load method to \"selectin\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/90dcbe1f6be04aba5ac90201045be3721c06625e"}]}],"author":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-01 15:34:15.000000000","tz":60},"committer":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-01 18:00:08.000000000","tz":60},"subject":"Fix python modules docs build","message":"Fix python modules docs build\n\nCurrently the Python code documentation is not being published at all\nwhich can be observed by visiting\n\nhttps://docs.openstack.org/neutron-lib/latest/genindex.html\n\nThis used to work in the old versions, for example in 1.9.1:\n\nhttps: //docs.openstack.org/neutron-lib/1.9.1/genindex.html\nChange-Id: Ibb919c53d7e1bc2710e85ae86a0ed0d7620a900c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cf30e80859702c317447bbb95f22ed7bacb1105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cf30e80859702c317447bbb95f22ed7bacb1105"}]},"branch":"refs/heads/master"},"f04bfd3e2c23e2bb656be289b9ce4c959ce40d9a":{"kind":"REWORK","_number":3,"created":"2024-08-14 15:44:13.000000000","uploader":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"ref":"refs/changes/03/925503/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/925503/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/3"}}},"commit":{"parents":[{"commit":"90dcbe1f6be04aba5ac90201045be3721c06625e","subject":"Merge \"Change the ``StandardAttribute`` load method to \"selectin\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/90dcbe1f6be04aba5ac90201045be3721c06625e"}]}],"author":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-01 15:34:15.000000000","tz":60},"committer":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-14 15:42:09.000000000","tz":60},"subject":"Fix python modules docs build","message":"Fix python modules docs build\n\nCurrently the Python code documentation is not being published at all\nwhich can be observed by visiting\n\nhttps://docs.openstack.org/neutron-lib/latest/genindex.html\n\nThis used to work in the old versions, for example in 1.9.1:\n\nhttps: //docs.openstack.org/neutron-lib/1.9.1/genindex.html\n\nDocstring fixes in placement/utils.py:\n\nWhen Sphinx generates documentation, it produces following warning:\n\"WARNING: Definition list ends without a blank line; unexpected unindent.\"\"\n\nThis commit adds the missing lines, but also switches to :: so that\ncontents of the definitions are treated as raw text.\n\nChange-Id: Ibb919c53d7e1bc2710e85ae86a0ed0d7620a900c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f04bfd3e2c23e2bb656be289b9ce4c959ce40d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f04bfd3e2c23e2bb656be289b9ce4c959ce40d9a"}]},"branch":"refs/heads/master"},"d69eae326e380440fb7acdf7c7a233eed43c3f32":{"kind":"REWORK","_number":4,"created":"2024-08-15 13:28:03.000000000","uploader":{"_account_id":37208,"name":"Marek Skrobacki","display_name":"Marek Skrobacki","email":"skrobul@skrobul.com","username":"skrobul"},"ref":"refs/changes/03/925503/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/925503/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/925503/4"}}},"commit":{"parents":[{"commit":"90dcbe1f6be04aba5ac90201045be3721c06625e","subject":"Merge \"Change the ``StandardAttribute`` load method to \"selectin\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/90dcbe1f6be04aba5ac90201045be3721c06625e"}]}],"author":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-01 15:34:15.000000000","tz":60},"committer":{"name":"skrobul","email":"skrobul@skrobul.com","date":"2024-08-15 13:26:52.000000000","tz":60},"subject":"Fix python modules docs build","message":"Fix python modules docs build\n\nCurrently the Python code documentation is not being published at all\nwhich can be observed by visiting\n\nhttps://docs.openstack.org/neutron-lib/latest/genindex.html\n\nThis used to work in the old versions, for example in 1.9.1:\n\nhttps: //docs.openstack.org/neutron-lib/1.9.1/genindex.html\n\nDocstring fixes in placement/utils.py:\n\nWhen Sphinx generates documentation, it produces following warning:\n\"WARNING: Definition list ends without a blank line; unexpected unindent.\"\"\n\nThis commit adds the missing lines, but also switches to :: so that\ncontents of the definitions are treated as raw text.\n\nChange-Id: Ibb919c53d7e1bc2710e85ae86a0ed0d7620a900c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d69eae326e380440fb7acdf7c7a233eed43c3f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d69eae326e380440fb7acdf7c7a233eed43c3f32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
